)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~664331","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~stable%2Fpike~I57853dfa418ae5c94e9efc45b72ac353e789571c","project":"openstack/openstack-ansible-os_gnocchi","branch":"stable/pike","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I57853dfa418ae5c94e9efc45b72ac353e789571c","subject":"Updated from OpenStack Ansible Tests","status":"ABANDONED","created":"2019-06-10 15:44:45.000000000","updated":"2020-10-02 12:26:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4e897ed32b861879fa44b40f033133f3e5744ba9","_number":664331,"virtual_id_number":664331,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-06-10 17:40:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 17:40:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"52c2c4178371a48f27239e83158bc69799e46687","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:44:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8517461d39e9430b2956552a0f0ea7926a137ec1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:40:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/31/664331/1/check/build-openstack-sphinx-docs/7dd5448/html/ : SUCCESS in 4m 14s\n- openstack-ansible-linters http://logs.openstack.org/31/664331/1/check/openstack-ansible-linters/076a1da/ : SUCCESS in 6m 30s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/31/664331/1/check/openstack-ansible-functional-centos-7/0b5a845/ : SUCCESS in 49m 27s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/31/664331/1/check/openstack-ansible-functional-opensuse-423/4b779f8/ : FAILURE in 9m 20s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/31/664331/1/check/openstack-ansible-functional-ubuntu-xenial/ce982cc/ : SUCCESS in 35m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1a9958d812348f4ab95de6f1f490bd990d87e141","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-02 12:26:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8400993e5bf20c2bfcb76705250ccf1a42b2a7c","revisions":{"e8400993e5bf20c2bfcb76705250ccf1a42b2a7c":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:44:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/664331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/31/664331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/31/664331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/31/664331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/31/664331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/31/664331/1"}}},"commit":{"parents":[{"commit":"7f3a396e379931c6d8ac66d77e96d88c1a29e5e8","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/7f3a396e379931c6d8ac66d77e96d88c1a29e5e8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:43.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I57853dfa418ae5c94e9efc45b72ac353e789571c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e8400993e5bf20c2bfcb76705250ccf1a42b2a7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e8400993e5bf20c2bfcb76705250ccf1a42b2a7c"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
