)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~670471","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Ic126b9ad6fa4905514f356b98690e04965f4d93c","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa/service-refactor","hashtags":[],"change_id":"Ic126b9ad6fa4905514f356b98690e04965f4d93c","subject":"service_setup: refactor service setup to a single file","status":"MERGED","created":"2019-07-12 08:15:54.000000000","updated":"2019-08-07 20:27:56.000000000","submitted":"2019-08-07 20:27:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670471-1565209676235-a273883a","meta_rev_id":"9d4063a06fe12989b4f98df4f841c2a3da22374d","_number":670471,"virtual_id_number":670471,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-07 16:21:14.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-08-07 20:27:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-02 12:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-30 20:30:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2019-08-02 12:57:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 20:30:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-02 12:57:52.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-08-07 20:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c09bd6b3f2e568bdb0b2031dd7c1db2df2d3f2e4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-12 08:15:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc84b2c3a0a0e86889bbb356a3b47f5101363317","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-12 08:16:46.000000000","message":"Patch Set 1: Workflow-1\n\nIt\u0027s supposingly fail with undefined _service_in_ldap","accounts_in_message":[],"_revision_number":1},{"id":"7b7cf0ce018c97c384b0eb436dc5f05cf12bd8b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 10:18:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/1/check/openstack-ansible-linters/c6fee59/ : SUCCESS in 6m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/1/check/openstack-ansible-deploy-aio_metal-debian-stable/4cff843/ : FAILURE in 51m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/1/check/openstack-ansible-deploy-aio_metal-centos-7/929e05d/ : FAILURE in 1h 07m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/6369c10/ : FAILURE in 59m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/41a7a13/ : FAILURE in 52m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/00d13f0/ : FAILURE in 50m 08s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/97b02c7/ : FAILURE in 46m 37s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/71/670471/1/check/openstack-tox-docs/b2e6e2a/html/ : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"edd6bcab25feb01681b4761ec4dc56ce7a3edf6c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-13 08:04:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"be32a6bb52399539ea97b1f3cf9473af8414ba80","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-13 08:06:45.000000000","message":"Patch Set 2:\n\nBE AWARE, that this patch has changes to service_setup.yml, so that after it\u0027s merge these changes are supposed to be manually distributed along roles https://review.opendev.org/#/q/topic:osa/service-refactor+status:merged","accounts_in_message":[],"_revision_number":2},{"id":"240358f9ac72502a76282577e2a398b7f53cfe3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-13 10:08:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/2/check/openstack-ansible-linters/18ab331/ : SUCCESS in 9m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/2/check/openstack-ansible-deploy-aio_metal-debian-stable/3e32d15/ : SUCCESS in 1h 25m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/2/check/openstack-ansible-deploy-aio_metal-centos-7/8bb0030/ : SUCCESS in 1h 38m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/6e8f34a/ : SUCCESS in 1h 38m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e4d7532/ : SUCCESS in 1h 22m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8bec7ad/ : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/6c14173/ : FAILURE in 50m 18s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/71/670471/2/check/openstack-tox-docs/ef70fde/html/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"b37e679232f3711d074efead8702b4b3853e6c4b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 16:44:13.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"46857dc9f2239f5eee1b93fcee384c696d4b324d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-18 17:24:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"90b5ee87fbb5200d61e372269877a43b223ac134","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 19:33:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/3/check/openstack-ansible-linters/4f5fea7/ : SUCCESS in 8m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/3/check/openstack-ansible-deploy-aio_metal-debian-stable/a84bed3/ : FAILURE in 1h 15m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/3/check/openstack-ansible-deploy-aio_metal-centos-7/fe97e67/ : FAILURE in 1h 19m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/ce4706a/ : FAILURE in 1h 21m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/72920cd/ : FAILURE in 1h 04m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/18dce48/ : FAILURE in 49m 54s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/f133054/ : FAILURE in 53m 37s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/71/670471/3/check/openstack-tox-docs/7d055f2/html/ : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":3},{"id":"957c86efb0fbbbd294c79381d2981d2107d2ee63","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 10:01:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3e05017f030596b66c5ec281559e946bab605867","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 11:55:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/4/check/openstack-ansible-linters/ded60e9/ : SUCCESS in 8m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_metal-debian-stable/0ea8d6a/ : SUCCESS in 1h 21m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_metal-centos-7/2a4e35b/ : SUCCESS in 1h 45m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_metal-opensuse-15/a6116f7/ : SUCCESS in 1h 48m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3d38ca0/ : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e6542ff/ : SUCCESS in 1h 23m 26s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/8dc9975/ : FAILURE in 58m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/1ccb5b0/ : FAILURE in 33m 50s\n- openstack-tox-docs http://logs.openstack.org/71/670471/4/check/openstack-tox-docs/a979238/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":4},{"id":"48ad59c73756b299dd9b0a3b4ff11ce9b5db2882","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 12:12:42.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"0ceab7fc090c15a01b61da2165fed54209fdfcc6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 12:17:27.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"0628b3f301cb620c42c59bca1522dbc7a4d94b29","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 13:45:01.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"74be3fcc5078afe9c733f29bca23db018ca95009","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 14:30:48.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/4/check/openstack-ansible-linters/c4d933d/ : SUCCESS in 9m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_metal-debian-stable/4ba37bf/ : FAILURE in 12m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_metal-centos-7/39d759a/ : SUCCESS in 2h 16m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_metal-opensuse-15/8fcc133/ : SUCCESS in 1h 32m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e205704/ : SUCCESS in 1h 35m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/92fa3a3/ : SUCCESS in 1h 11m 26s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/b16ed21/ : FAILURE in 1h 11m 56s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/670471/4/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/38f6c7c/ : FAILURE in 37m 59s\n- openstack-tox-docs http://logs.openstack.org/71/670471/4/check/openstack-tox-docs/2c82b78/html/ : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":4},{"id":"c3095498a41b746686f3fe875755924a1183b322","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-22 10:07:40.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"4ee235ba6fcb473dd6a182f7341c89959b97a2a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 11:31:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/5/check/openstack-ansible-linters/2995e63/ : SUCCESS in 5m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/5/check/openstack-ansible-deploy-aio_metal-debian-stable/48706fa/ : FAILURE in 59m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/5/check/openstack-ansible-deploy-aio_metal-centos-7/af1aff3/ : FAILURE in 1h 16m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/5/check/openstack-ansible-deploy-aio_metal-opensuse-15/bf72749/ : FAILURE in 1h 10m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/5/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f3080c5/ : FAILURE in 52m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/5/check/openstack-ansible-deploy-aio_distro_metal-centos-7/4092d37/ : FAILURE in 59m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/5/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/a344b40/ : FAILURE in 56m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/670471/5/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/8172def/ : FAILURE in 48m 01s\n- openstack-tox-docs http://logs.openstack.org/71/670471/5/check/openstack-tox-docs/166f457/html/ : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":5},{"id":"e626c5e0758c3c4e591fb15b8d71a22f06f992f9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-25 12:08:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b928212449aa9a1c111e058eb6216e749087c335","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 14:06:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/6/check/openstack-ansible-linters/2df320a/ : SUCCESS in 7m 26s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/6/check/openstack-ansible-deploy-aio_metal-debian-stable/6550bcc/ : SUCCESS in 1h 13m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/6/check/openstack-ansible-deploy-aio_metal-centos-7/065ada4/ : SUCCESS in 1h 50m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/6/check/openstack-ansible-deploy-aio_metal-opensuse-15/e91d47c/ : SUCCESS in 1h 36m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/6/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fe94c9d/ : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/6/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e312c04/ : SUCCESS in 1h 18m 55s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/6/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/cf44d2c/ : FAILURE in 51m 19s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/670471/6/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/7bc2872/ : FAILURE in 40m 59s\n- openstack-tox-docs http://logs.openstack.org/71/670471/6/check/openstack-tox-docs/5be65a3/html/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":6},{"id":"269d449281a8a13305d47db21f379d38d858da01","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-25 20:02:43.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"ae1f66311f6b87e60ce0aef3c42d2fe237b0cd1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 22:02:32.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/7/check/openstack-ansible-linters/c769139/ : SUCCESS in 6m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-debian-stable/4aeaf0a/ : SUCCESS in 1h 19m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-centos-7/f742125/ : SUCCESS in 1h 38m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-opensuse-15/2162a3a/ : SUCCESS in 1h 28m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d4aa95d/ : SUCCESS in 1h 08m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-centos-7/07e7f7d/ : SUCCESS in 1h 17m 55s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/fc53b56/ : FAILURE in 45m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/57f05af/ : FAILURE in 38m 38s\n- openstack-tox-docs http://logs.openstack.org/71/670471/7/check/openstack-tox-docs/76f412d/html/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":7},{"id":"b7181c20d1d59202107b384087f74d7d28228b6a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-26 05:11:24.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"58f7f2be7cebae963d0463987ddd379e304bb309","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 06:31:23.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/7/check/openstack-ansible-linters/eec7854/ : SUCCESS in 5m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-debian-stable/d032c73/ : FAILURE in 53m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-centos-7/b388fcc/ : FAILURE in 1h 15m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-opensuse-15/7925dbe/ : FAILURE in 53m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/a1ea813/ : FAILURE in 46m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-centos-7/a201f1a/ : FAILURE in 1h 01m 12s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/006d4f2/ : FAILURE in 38m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/fbe73f8/ : FAILURE in 37m 25s\n- openstack-tox-docs http://logs.openstack.org/71/670471/7/check/openstack-tox-docs/f5260ee/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":7},{"id":"0e517452d8566489cb12bd44924999bb8fe65aec","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-26 19:48:42.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"fa5fd012d9957e2346658d9e748265a9341a5d4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 21:05:49.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/7/check/openstack-ansible-linters/6f01225/ : SUCCESS in 9m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-debian-stable/386b1da/ : FAILURE in 45m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-centos-7/a0c045e/ : FAILURE in 1h 01m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-opensuse-15/2b4d650/ : FAILURE in 1h 02m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/920b5e5/ : FAILURE in 49m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-centos-7/0c354b7/ : FAILURE in 45m 41s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/37289e1/ : FAILURE in 44m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/3bde55e/ : FAILURE in 33m 28s\n- openstack-tox-docs http://logs.openstack.org/71/670471/7/check/openstack-tox-docs/ac4b7c4/html/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":7},{"id":"ad38e03ceba08434e300fdc9f457a8a506562d6b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-27 08:24:46.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"98de75ed8ddfdc318eda05cc174e2aadf9fb31b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 10:07:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/7/check/openstack-ansible-linters/3ea2cc4/ : SUCCESS in 5m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-debian-stable/baadf8a/ : SUCCESS in 1h 21m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-centos-7/85c38cd/ : SUCCESS in 1h 37m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-opensuse-15/eeb3b38/ : SUCCESS in 1h 31m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/c93b4aa/ : SUCCESS in 1h 20m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-centos-7/ddc7d31/ : SUCCESS in 1h 27m 56s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/6c16819/ : FAILURE in 49m 37s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/670471/7/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/67eb008/ : SUCCESS in 56m 47s\n- openstack-tox-docs http://logs.openstack.org/71/670471/7/check/openstack-tox-docs/f3dde56/html/ : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":7},{"id":"da7b67fdfd325b73c48bd65b816152ab145ffdcf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-27 20:40:24.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"7a306d434a057af6be22e8c15d35523b040677d5","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-29 20:49:09.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"cc8522d6181add8d2fb987d643a77899031d2f9e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 05:55:49.000000000","message":"Patch Set 7: Workflow+1\n\npush to the gate","accounts_in_message":[],"_revision_number":7},{"id":"e1faad7ab77f0f7fb1e908e57777a706ab0d28ff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 11:11:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"60ba2345a34ddd6ea8af94661d28c558cfd4adb1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 11:18:15.000000000","message":"Topic changed from osa/servce_refactor to osa/service_refactor","accounts_in_message":[],"_revision_number":8},{"id":"a9858b12d73cc529f293557670dcac793f0ac8c8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 11:18:53.000000000","message":"Topic changed from osa/service_refactor to osa/service-refactor","accounts_in_message":[],"_revision_number":8},{"id":"48d450ad607cac46c0cf04655b099addef093dfb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 11:37:29.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"cf0625edbbf8666e6168772a026636cef6c07ecf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 13:25:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/71/670471/9/check/openstack-ansible-linters/6804f58/ : SUCCESS in 7m 13s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-debian-stable/6f66886/ : SUCCESS in 1h 23m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-centos-7/c8d9dce/ : SUCCESS in 1h 45m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-opensuse-15/9824379/ : SUCCESS in 1h 31m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b35ea04/ : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5dfc451/ : SUCCESS in 1h 20m 42s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/b27056e/ : FAILURE in 47m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/a6a3c16/ : SUCCESS in 56m 57s\n- openstack-tox-docs http://logs.openstack.org/71/670471/9/check/openstack-tox-docs/421393c/html/ : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":9},{"id":"be68206bfe7d29da2119ccbbdcd39b1aea164447","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-30 20:30:35.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"8465fee82fe8564ecb85b02ce79594f4caff72a9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-08-02 12:57:52.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"984b2ae2a942d050d3123236167439e4a70c0016","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:58:04.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"480607855b82b45052099d82dcb646f08fb540b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 14:31:04.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/gate/openstack-ansible-linters/38a439f/ : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-debian-stable/d53c916/ : FAILURE in 1h 13m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-centos-7/e4f66e9/ : FAILURE in 1h 27m 27s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-opensuse-15/9b0fe37/ : FAILURE in 1h 11m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/31f79c3/ : FAILURE in 1h 05m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/989928f/ : FAILURE in 1h 18m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/ad34027/ : FAILURE in 42m 28s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/gate/openstack-tox-docs/1f35eb7/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":9},{"id":"40589099b2396f7a7841f9b89af1cce6f1cfc004","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-02 15:27:58.000000000","message":"Patch Set 9:\n\nhttps://review.opendev.org/#/c/674295/ Should resolve the problem","accounts_in_message":[],"_revision_number":9},{"id":"2c1ea18f1cb57e828a94e8bc5f1d24d4f5d7127a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-05 12:00:30.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"d9bd6cfaa13cf362ada44f9f2764d9ed0f3085b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:00:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/check/openstack-ansible-linters/692ce50/ : SUCCESS in 8m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-debian-stable/1c6ac56/ : SUCCESS in 1h 28m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-centos-7/cc1e21a/ : SUCCESS in 1h 42m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-opensuse-15/5f53adc/ : SUCCESS in 1h 58m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3169451/ : FAILURE in 1h 31m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/f969e16/ : SUCCESS in 1h 14m 49s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/21972ae/ : FAILURE in 45m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/06256c7/ : SUCCESS in 1h 15m 24s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/check/openstack-tox-docs/3786bcc/html/ : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":9},{"id":"2b5c3e16bb61095ba11c1bc9fe2fced800bc936f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-05 15:49:30.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"3c24f171ba50814b3a7f24484a52336f9e02f587","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 18:02:29.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/check/openstack-ansible-linters/abd395c/ : SUCCESS in 9m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-debian-stable/5447a15/ : SUCCESS in 1h 28m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-centos-7/b1e9140/ : SUCCESS in 1h 49m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-opensuse-15/d833344/ : RETRY_LIMIT in 8m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dc05cb7/ : SUCCESS in 1h 03m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/434e31e/ : SUCCESS in 1h 35m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/b57b5ad/ : FAILURE in 48m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/0a0771e/ : SUCCESS in 53m 47s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/check/openstack-tox-docs/8198406/html/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":9},{"id":"947d1830e09aa6e9e6f7fb83a7f80091b193527f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-06 10:12:50.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"f37f04889e4d92f2aec217acf5585623db418d7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:36:09.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/check/openstack-ansible-linters/55a5c88/ : SUCCESS in 6m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-debian-stable/a37c10c/ : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-centos-7/bdb630b/ : SUCCESS in 2h 15m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-opensuse-15/1106455/ : SUCCESS in 1h 21m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8693c5d/ : SUCCESS in 1h 14m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/8c1edd9/ : SUCCESS in 1h 34m 36s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/983c10d/ : FAILURE in 58m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/80727b9/ : SUCCESS in 58m 36s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/check/openstack-tox-docs/21bf913/html/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":9},{"id":"9ddaae1e37b7d095b5ea2db8a43945482c16b6ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 12:36:22.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"125132b8ecf52cb5db0b9ff9518e18630eb4bb09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 15:43:16.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/gate/openstack-ansible-linters/301e72e/ : SUCCESS in 4m 18s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-debian-stable/915772b/ : SUCCESS in 1h 18m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-centos-7/dc853f9/ : SUCCESS in 2h 17m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-opensuse-15/ceb41ad/ : SUCCESS in 1h 40m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8efdf0f/ : TIMED_OUT in 3h 03m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/ffe2aac/ : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/6d49469/ : SUCCESS in 52m 25s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/gate/openstack-tox-docs/1c7f0e9/html/ : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":9},{"id":"53686178cead7600a3b62c268312042de13b2d95","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-06 16:30:58.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"0f6b465ebfdfb0606bda8959483d68d0a38f1f00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 18:27:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/check/openstack-ansible-linters/da6878d/ : SUCCESS in 5m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-debian-stable/6bfc235/ : SUCCESS in 1h 23m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-centos-7/22686b7/ : SUCCESS in 1h 47m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-opensuse-15/9f54dd4/ : FAILURE in 1h 17m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/64cf0e9/ : SUCCESS in 1h 14m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/eaa18c7/ : SUCCESS in 1h 11m 38s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/f9f08fa/ : FAILURE in 47m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/176ad15/ : SUCCESS in 57m 49s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/check/openstack-tox-docs/78ad745/html/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":9},{"id":"ac3023aa33ff73734bb7431a1deaa1a3cd605b7b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-06 18:53:14.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"2c8b20d0b5c79815e31871d87b9d24246f9350bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 20:44:02.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/check/openstack-ansible-linters/db2a494/ : SUCCESS in 8m 31s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-debian-stable/f988bff/ : SUCCESS in 1h 13m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-centos-7/c0a3add/ : SUCCESS in 1h 42m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-opensuse-15/85960d7/ : FAILURE in 1h 18m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b44bd27/ : SUCCESS in 1h 16m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/cf1a8e3/ : SUCCESS in 1h 13m 35s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/0a25b53/ : FAILURE in 54m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/0461b3b/ : SUCCESS in 1h 09m 59s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/check/openstack-tox-docs/bcdc290/html/ : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":9},{"id":"8156e24e4202e5ad9bb6bf060aa21811c09d9e78","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-07 11:09:34.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e01f38c22aa45d390017ec5a45e575fcd6a418db","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 14:17:20.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/check/openstack-ansible-linters/11d95ea/ : SUCCESS in 6m 47s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-debian-stable/7d9de2f/ : SUCCESS in 1h 25m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-centos-7/58433ad/ : SUCCESS in 1h 51m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-opensuse-15/3ffaf5d/ : SUCCESS in 1h 25m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/467c7cd/ : TIMED_OUT in 3h 04m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/82d54ca/ : SUCCESS in 1h 18m 58s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/8930c9b/ : FAILURE in 52m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/6c6ba53/ : SUCCESS in 52m 12s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/check/openstack-tox-docs/73ac98d/html/ : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":9},{"id":"bdc97c883b3561fcedd0d719a650a6d91892237e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-07 16:21:14.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"15c48bea9eb69f98b5c9bbc3d40e1526fd121f2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 18:26:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/check/openstack-ansible-linters/c685aea/ : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-debian-stable/01405aa/ : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-centos-7/d9d03b0/ : SUCCESS in 1h 58m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-opensuse-15/8f27cbd/ : SUCCESS in 1h 45m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/25cd1de/ : SUCCESS in 1h 28m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-centos-7/976b34f/ : SUCCESS in 1h 14m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/6eddc2f/ : FAILURE in 48m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/488eb71/ : SUCCESS in 59m 30s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/check/openstack-tox-docs/7b03f25/html/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":9},{"id":"c90fcc53843a56fc00b6d106ae721519758f2011","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 18:26:17.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"632f116773682f6d06ae52a7e08c8c85e10d055f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 20:27:55.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://logs.opendev.org/71/670471/9/gate/openstack-ansible-linters/6910006/ : SUCCESS in 5m 26s\n- openstack-ansible-deploy-aio_metal-debian-stable https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-debian-stable/707c3dc/ : SUCCESS in 1h 24m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-centos-7/cc68e5d/ : SUCCESS in 1h 55m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-opensuse-15/a3aec40/ : SUCCESS in 1h 52m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e6dee13/ : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/10f8b64/ : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://logs.opendev.org/71/670471/9/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/35430a0/ : SUCCESS in 51m 22s\n- openstack-tox-docs https://logs.opendev.org/71/670471/9/gate/openstack-tox-docs/33618ca/html/ : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":9},{"id":"9d4063a06fe12989b4f98df4f841c2a3da22374d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 20:27:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"17a2175bdadef30e582ccbb5ee4a09f46f4b7c95","revisions":{"4f3d5e12de26cdb36725747064721e916239b5a9":{"kind":"REWORK","_number":1,"created":"2019-07-12 08:15:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/670471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/670471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/1"}}},"commit":{"parents":[{"commit":"d3d9b327ec28c87c9419d313dcf8502aebb3b670","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d3d9b327ec28c87c9419d313dcf8502aebb3b670"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: Ic126b9ad6fa4905514f356b98690e04965f4d93c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4f3d5e12de26cdb36725747064721e916239b5a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4f3d5e12de26cdb36725747064721e916239b5a9"}]},"branch":"refs/heads/master"},"0ab87ecd975543dee277368ed7a0f06b78f794d8":{"kind":"REWORK","_number":2,"created":"2019-07-13 08:04:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/670471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/670471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/2"}}},"commit":{"parents":[{"commit":"d3d9b327ec28c87c9419d313dcf8502aebb3b670","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d3d9b327ec28c87c9419d313dcf8502aebb3b670"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-13 08:03:58.000000000","tz":180},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: Ic126b9ad6fa4905514f356b98690e04965f4d93c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/0ab87ecd975543dee277368ed7a0f06b78f794d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/0ab87ecd975543dee277368ed7a0f06b78f794d8"}]},"branch":"refs/heads/master"},"2211099caaf871edb6660c3d645ac838070fe6ea":{"kind":"REWORK","_number":3,"created":"2019-07-18 17:24:14.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/670471/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/670471/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/3"}}},"commit":{"parents":[{"commit":"d3d9b327ec28c87c9419d313dcf8502aebb3b670","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d3d9b327ec28c87c9419d313dcf8502aebb3b670"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-18 17:24:00.000000000","tz":180},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: Ic126b9ad6fa4905514f356b98690e04965f4d93c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2211099caaf871edb6660c3d645ac838070fe6ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2211099caaf871edb6660c3d645ac838070fe6ea"}]},"branch":"refs/heads/master"},"4123c9363384c9f7a59df7f13d12b9ded7e2b755":{"kind":"REWORK","_number":4,"created":"2019-07-19 10:01:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/670471/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/670471/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/4"}}},"commit":{"parents":[{"commit":"d3d9b327ec28c87c9419d313dcf8502aebb3b670","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d3d9b327ec28c87c9419d313dcf8502aebb3b670"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-19 10:01:33.000000000","tz":180},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: Ic126b9ad6fa4905514f356b98690e04965f4d93c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4123c9363384c9f7a59df7f13d12b9ded7e2b755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4123c9363384c9f7a59df7f13d12b9ded7e2b755"}]},"branch":"refs/heads/master"},"263e0f4903728f3a5bbdc91c555ee4e34000dbcb":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-22 10:07:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/670471/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/670471/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/5"}}},"commit":{"parents":[{"commit":"d3d9b327ec28c87c9419d313dcf8502aebb3b670","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d3d9b327ec28c87c9419d313dcf8502aebb3b670"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-07-22 10:07:36.000000000","tz":0},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nDepends-On: https://review.opendev.org/671988\nChange-Id: Ic126b9ad6fa4905514f356b98690e04965f4d93c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/263e0f4903728f3a5bbdc91c555ee4e34000dbcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/263e0f4903728f3a5bbdc91c555ee4e34000dbcb"}]},"branch":"refs/heads/master"},"433c840d127d4df107df809d8a6c4b0f2f46be2b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2019-07-25 12:08:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/670471/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/670471/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/6"}}},"commit":{"parents":[{"commit":"f4e68f14909be9de84ffe955fb450fce3a2c2acb","subject":"Fix gnocchi wsgi scrpt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f4e68f14909be9de84ffe955fb450fce3a2c2acb"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-25 12:08:09.000000000","tz":180},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: Ic126b9ad6fa4905514f356b98690e04965f4d93c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/433c840d127d4df107df809d8a6c4b0f2f46be2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/433c840d127d4df107df809d8a6c4b0f2f46be2b"}]},"branch":"refs/heads/master"},"171aeb0a6af70e354546d727304dddf55e7d77f9":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-07-25 20:02:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/670471/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/670471/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/7"}}},"commit":{"parents":[{"commit":"f4e68f14909be9de84ffe955fb450fce3a2c2acb","subject":"Fix gnocchi wsgi scrpt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f4e68f14909be9de84ffe955fb450fce3a2c2acb"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-07-25 20:02:36.000000000","tz":0},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nDepends-On: https://review.opendev.org/671988\nChange-Id: Ic126b9ad6fa4905514f356b98690e04965f4d93c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/171aeb0a6af70e354546d727304dddf55e7d77f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/171aeb0a6af70e354546d727304dddf55e7d77f9"}]},"branch":"refs/heads/master"},"e9d72c15250a736c60ede0c56a16d4ca4b3f1c65":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":8,"created":"2019-07-30 11:11:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/670471/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/670471/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/8"}}},"commit":{"parents":[{"commit":"57b88ebc65e516322a0268509816ec1b8640b1db","subject":"Migrate gnocchi to use uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/57b88ebc65e516322a0268509816ec1b8640b1db"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-30 11:09:32.000000000","tz":180},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: Ic126b9ad6fa4905514f356b98690e04965f4d93c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e9d72c15250a736c60ede0c56a16d4ca4b3f1c65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e9d72c15250a736c60ede0c56a16d4ca4b3f1c65"}]},"branch":"refs/heads/master"},"17a2175bdadef30e582ccbb5ee4a09f46f4b7c95":{"kind":"REWORK","_number":9,"created":"2019-07-30 11:37:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/670471/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/670471/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/670471/9"}}},"commit":{"parents":[{"commit":"57b88ebc65e516322a0268509816ec1b8640b1db","subject":"Migrate gnocchi to use uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/57b88ebc65e516322a0268509816ec1b8640b1db"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-07-12 08:15:37.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-07-30 11:36:55.000000000","tz":0},"subject":"service_setup: refactor service setup to a single file","message":"service_setup: refactor service setup to a single file\n\nThis patch refactors the openstack user/service/endpoints creation to\nservice_setup.yml which will eventually be managed by\nopenstack-ansible-tests.\n\nChange-Id: Ic126b9ad6fa4905514f356b98690e04965f4d93c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/17a2175bdadef30e582ccbb5ee4a09f46f4b7c95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/17a2175bdadef30e582ccbb5ee4a09f46f4b7c95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
