)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~671123","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I3f6a1cee0e98372881b015ebf06e405c79495fe1","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"use_journalctl","hashtags":[],"change_id":"I3f6a1cee0e98372881b015ebf06e405c79495fe1","subject":"Use systemd-journald instead of log files","status":"MERGED","created":"2019-07-16 17:46:46.000000000","updated":"2019-07-17 07:15:16.000000000","submitted":"2019-07-17 07:15:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671123-1563347716903-0a1ee91c","meta_rev_id":"49d2a6b8f7150eafa218952c093055296138b313","_number":671123,"virtual_id_number":671123,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-17 07:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-16 21:35:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-17 05:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-17 05:16:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 21:35:19.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-17 05:16:17.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-17 07:15:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4471e36c4224b8fd310ba7bdbfb3a63dbec775b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 17:46:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8c95f6b7c34cb1bc423faa8b437bb366ba6631d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-16 19:23:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3cdf8b7b850c7051ae971e64c5d846d85228911a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 19:26:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/23/671123/1/check/openstack-ansible-linters/4a4da63/ : SUCCESS in 7m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/23/671123/1/check/openstack-ansible-deploy-aio_metal-debian-stable/1d38023/ : SUCCESS in 1h 34m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/23/671123/1/check/openstack-ansible-deploy-aio_metal-centos-7/3afc10b/ : SUCCESS in 1h 38m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/23/671123/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/724fefc/ : SUCCESS in 1h 33m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/23/671123/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b4e20c2/ : SUCCESS in 1h 14m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/23/671123/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3aca2a1/ : SUCCESS in 1h 16m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/23/671123/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/a7431dd/ : FAILURE in 53m 40s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/23/671123/1/check/openstack-tox-docs/aebaf75/html/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8b20d5c7239a2a5507b9bad79e3c14dae8d493b0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 21:26:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"26a9d085555fca759f8f75aa39a90b58948fe579","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-16 21:26:49.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d39b155fbcba8644715b27715a93d89b27a0687f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-16 21:35:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a3d43e64934f7bbc0635eda3c4fb920d3f49764","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:14:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/23/671123/2/check/openstack-ansible-linters/b543cfe/ : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/23/671123/2/check/openstack-ansible-deploy-aio_metal-debian-stable/32edd34/ : SUCCESS in 1h 08m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/23/671123/2/check/openstack-ansible-deploy-aio_metal-centos-7/2ae290b/ : SUCCESS in 1h 42m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/23/671123/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/63c28ea/ : SUCCESS in 1h 22m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/23/671123/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d4fe668/ : SUCCESS in 1h 03m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/23/671123/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/fc6a0ae/ : SUCCESS in 1h 29m 24s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/23/671123/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/2b83101/ : FAILURE in 46m 48s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/23/671123/2/check/openstack-tox-docs/ffb797a/html/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"58a27f56a4149109638aaeecd57b19508f740cd9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-17 05:16:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8924f86de6065670a77471983eb8311c033be16a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 05:16:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"628381189034ae0cc9b4986770bd3a69b1327b88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 07:15:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/23/671123/2/gate/openstack-ansible-linters/7c749b8/ : SUCCESS in 4m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/23/671123/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/4a89188/ : SUCCESS in 1h 19m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/23/671123/2/gate/openstack-ansible-deploy-aio_metal-centos-7/4cd1977/ : SUCCESS in 1h 53m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/23/671123/2/gate/openstack-ansible-deploy-aio_metal-opensuse-15/c063d5f/ : SUCCESS in 1h 23m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/23/671123/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/63c80cc/ : SUCCESS in 1h 08m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/23/671123/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/5ee6001/ : SUCCESS in 1h 10m 45s\n- openstack-tox-docs http://logs.openstack.org/23/671123/2/gate/openstack-tox-docs/53d118e/html/ : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":2},{"id":"10086cf7801a0bb847e445b1c3016953ad11c94e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 07:15:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"eb083ecf9f150cd1d42b31c8bfd17b5038ba004a","revisions":{"b83d6a4b73e9dd4faa868e2507bad001019e9c59":{"kind":"REWORK","_number":1,"created":"2019-07-16 17:46:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/23/671123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/23/671123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/23/671123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/23/671123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/23/671123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/23/671123/1"}}},"commit":{"parents":[{"commit":"426492fbc9b68afc0a9646edbf0c118742150b3d","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/426492fbc9b68afc0a9646edbf0c118742150b3d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 17:46:34.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 17:46:34.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald. We also disable uwsgi logging, since it dublicates\nrequests that are logged by service itself.\n\nChange-Id: I3f6a1cee0e98372881b015ebf06e405c79495fe1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b83d6a4b73e9dd4faa868e2507bad001019e9c59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b83d6a4b73e9dd4faa868e2507bad001019e9c59"}]},"branch":"refs/heads/master"},"eb083ecf9f150cd1d42b31c8bfd17b5038ba004a":{"kind":"REWORK","_number":2,"created":"2019-07-16 21:26:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/23/671123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/23/671123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/23/671123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/23/671123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/23/671123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/23/671123/2"}}},"commit":{"parents":[{"commit":"426492fbc9b68afc0a9646edbf0c118742150b3d","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/426492fbc9b68afc0a9646edbf0c118742150b3d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 17:46:34.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-16 21:26:00.000000000","tz":180},"subject":"Use systemd-journald instead of log files","message":"Use systemd-journald instead of log files\n\nThis patch aims to migrate service from usage of regular syslog files\nto journald. We also disable uwsgi logging, since it dublicates\nrequests that are logged by service itself.\n\nChange-Id: I3f6a1cee0e98372881b015ebf06e405c79495fe1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/eb083ecf9f150cd1d42b31c8bfd17b5038ba004a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/eb083ecf9f150cd1d42b31c8bfd17b5038ba004a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
