)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~671599","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I2db6ad400d743615a18bcd533228e20d5402970a","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I2db6ad400d743615a18bcd533228e20d5402970a","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-18 22:18:20.000000000","updated":"2019-08-23 06:15:22.000000000","submitted":"2019-08-23 06:13:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671599-1566540787292-1b6a3505","meta_rev_id":"dc152698bd24b9700783c147d6c590a1a28aced3","_number":671599,"virtual_id_number":671599,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-08-23 06:13:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-08-20 17:44:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-08-20 17:44:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 22:05:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-08-20 17:44:31.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-08-23 06:13:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6666d2b23164b8c8d0f9e1a29cb56ea6f4de41f9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-18 22:18:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdf15af15e6d732b9d47a67bd42d8cd01fe1c559","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 00:12:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/99/671599/1/check/openstack-ansible-linters/8af129b/ : SUCCESS in 6m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/99/671599/1/check/openstack-ansible-deploy-aio_metal-debian-stable/dc79c76/ : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/99/671599/1/check/openstack-ansible-deploy-aio_metal-centos-7/c879d87/ : SUCCESS in 1h 33m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/99/671599/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/dba9e88/ : SUCCESS in 1h 30m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/99/671599/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/61b8dc7/ : SUCCESS in 1h 13m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/99/671599/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/da20c83/ : SUCCESS in 1h 18m 01s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/99/671599/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/8442883/ : FAILURE in 47m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/99/671599/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/a915d1b/ : FAILURE in 48m 41s\n- openstack-tox-docs http://logs.openstack.org/99/671599/1/check/openstack-tox-docs/172969a/html/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"112c07618b5143cad8ad5ff6983dd2d7e0db04e1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-19 18:21:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95d9feafb5e2d8de2e8beebc9dcf2299fa3164dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 20:56:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/99/671599/2/check/requirements-check/effb48a/ : SUCCESS in 2m 14s\n- openstack-ansible-linters http://logs.openstack.org/99/671599/2/check/openstack-ansible-linters/b6b9db8/ : SUCCESS in 8m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/99/671599/2/check/openstack-ansible-deploy-aio_metal-debian-stable/087dc9a/ : SUCCESS in 1h 13m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/99/671599/2/check/openstack-ansible-deploy-aio_metal-centos-7/22a375f/ : SUCCESS in 2h 04m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/99/671599/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/d6feb77/ : SUCCESS in 2h 00m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/99/671599/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0b6307b/ : SUCCESS in 1h 20m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/99/671599/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/ea5bb47/ : SUCCESS in 1h 17m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/99/671599/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/59d38ff/ : FAILURE in 49m 54s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/99/671599/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/2a86e3c/ : FAILURE in 50m 05s\n- openstack-tox-docs http://logs.openstack.org/99/671599/2/check/openstack-tox-docs/fc499ec/ : POST_FAILURE in 3m 26s\n- build-openstack-releasenotes http://logs.openstack.org/99/671599/2/check/build-openstack-releasenotes/8dac447/html/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d7196f696f216ffd7a050fe0bb8e90675416e2b8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-20 09:32:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"707e601c8d938d3ed9b567d9f08a8ea86da7582a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 11:19:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/99/671599/3/check/requirements-check/990d4ea/ : SUCCESS in 2m 56s\n- openstack-ansible-linters http://logs.openstack.org/99/671599/3/check/openstack-ansible-linters/67f6791/ : SUCCESS in 8m 13s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/99/671599/3/check/openstack-ansible-deploy-aio_metal-debian-stable/379ed72/ : SUCCESS in 1h 18m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/99/671599/3/check/openstack-ansible-deploy-aio_metal-centos-7/9382fae/ : SUCCESS in 1h 41m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/99/671599/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/9e3b059/ : SUCCESS in 1h 25m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/99/671599/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d1abac1/ : SUCCESS in 1h 06m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/99/671599/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/ae0133a/ : SUCCESS in 1h 18m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/99/671599/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/024545c/ : FAILURE in 43m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/99/671599/3/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/0bf41a0/ : FAILURE in 38m 06s\n- openstack-tox-docs http://logs.openstack.org/99/671599/3/check/openstack-tox-docs/0af138d/html/ : SUCCESS in 2m 30s\n- build-openstack-releasenotes http://logs.openstack.org/99/671599/3/check/build-openstack-releasenotes/26b9d7b/html/ : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":3},{"id":"612f7403f86ef1b5aec0aa7de1103d1dfe33370d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-30 19:39:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c35285eb36ca301acbeebd772615a21713d3867f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 21:51:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/99/671599/4/check/requirements-check/d037a5c/ : SUCCESS in 3m 34s\n- openstack-ansible-linters http://logs.openstack.org/99/671599/4/check/openstack-ansible-linters/c707ea5/ : SUCCESS in 8m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/99/671599/4/check/openstack-ansible-deploy-aio_metal-debian-stable/ace6b93/ : SUCCESS in 1h 18m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/99/671599/4/check/openstack-ansible-deploy-aio_metal-centos-7/4157f59/ : SUCCESS in 1h 53m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/99/671599/4/check/openstack-ansible-deploy-aio_metal-opensuse-15/cc3e102/ : SUCCESS in 1h 36m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/99/671599/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/7094337/ : SUCCESS in 1h 11m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/99/671599/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/b887b14/ : SUCCESS in 1h 24m 03s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/99/671599/4/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/a00805b/ : FAILURE in 52m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/99/671599/4/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/8e0a657/ : SUCCESS in 53m 34s\n- openstack-tox-docs http://logs.openstack.org/99/671599/4/check/openstack-tox-docs/63e306a/html/ : SUCCESS in 2m 59s\n- build-openstack-releasenotes http://logs.openstack.org/99/671599/4/check/build-openstack-releasenotes/5ceee3d/html/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"9de26bd165e37a65db7020611b3bec10b17968f6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-30 22:05:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5c305748bdc06316263274f7a2612dfe6c91d280","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 22:05:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"70f022a7233ccb7551ca4a4f366eddf8132acd49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 23:41:36.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/99/671599/4/gate/requirements-check/e838c60/ : SUCCESS in 2m 06s\n- openstack-ansible-linters http://logs.openstack.org/99/671599/4/gate/openstack-ansible-linters/6025469/ : SUCCESS in 6m 11s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/99/671599/4/gate/openstack-ansible-deploy-aio_metal-debian-stable/1cf642e/ : FAILURE in 1h 11m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/99/671599/4/gate/openstack-ansible-deploy-aio_metal-centos-7/bcc1796/ : FAILURE in 1h 27m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/99/671599/4/gate/openstack-ansible-deploy-aio_metal-opensuse-15/9fb4950/ : FAILURE in 1h 12m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/99/671599/4/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d68b29d/ : FAILURE in 59m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/99/671599/4/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/4dead9a/ : FAILURE in 1h 04m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/99/671599/4/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/c0facdf/ : FAILURE in 54m 25s\n- openstack-tox-docs http://logs.openstack.org/99/671599/4/gate/openstack-tox-docs/e110350/html/ : SUCCESS in 3m 28s\n- build-openstack-releasenotes http://logs.openstack.org/99/671599/4/gate/build-openstack-releasenotes/b28b3a2/html/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":4},{"id":"627ecb4bc82b7558977f20c7d8899f58ced8e970","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-19 22:20:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"faf63995422bc37124548ea88e4a85afb448e9c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 00:19:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e4c07b3d784746ce842e65ad439fec0c : SUCCESS in 3m 38s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b839e7598edd400faf6d74ad47c05464 : SUCCESS in 5m 52s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/89f25ea47458452d90b3acd9a9dd66eb : SUCCESS in 1h 34m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a1cc16abbea74543ae132035a641c1f3 : SUCCESS in 1h 48m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/072ac667c1884ac88f5fff8f306d02e0 : SUCCESS in 1h 25m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4f305e6246e46bba537cc9907a1deb4 : SUCCESS in 1h 29m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3fcd82f415c243c993cb09bff492657e : SUCCESS in 1h 17m 52s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9012d0eb6ce14c02bd8f9d1a6ce7ec4e : FAILURE in 46m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc50c4daa9d148a38df57ad0d4ee2118 : SUCCESS in 55m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08876a7ebbca417da21b07f484aed6c1 : SUCCESS in 4m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1290fbd340ac447d96263aa789cad0b6 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":5},{"id":"a4498e7a0e73fef59045ab3cf2e8564b09fcef15","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-08-20 03:06:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dbec84aa00bdf9091e7bec32dde9faf53e829e71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 04:49:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e8f56a455ba49639134c7b65be79fc5 : SUCCESS in 3m 58s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c46bf3626f60466aba3bf2ca71c5d480 : SUCCESS in 9m 41s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e289326b44e74b319fd118da335b01cf : FAILURE in 1h 13m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bdbd040c37fa427eba0d731ce1bcc9fb : SUCCESS in 1h 41m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/edd31d57d7a14f09a53807c48ed24ee1 : FAILURE in 1h 23m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61a86d8749fb411c8b3cf307d3ee6595 : FAILURE in 1h 22m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/50788dfbc08142869d57483ee57669b5 : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e1ff6263d440402faf5fee35e95ed0e6 : FAILURE in 55m 51s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/823fb74543b94597b22c463be4086cfb : FAILURE in 54m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287c7f01fba4412d9dccd93e9b859d9c : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fecdb13661f435dbcb16d4d0dac0176 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":6},{"id":"7e87b5d345390948604ac5225e63965292fc499a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-20 11:03:12.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"149250dc21ce1303ca39a9e86e8af22d7759a34f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 13:00:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/330365845fdf4ec9a4de5e42c9a5bdff : SUCCESS in 2m 44s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/09ec05b4aeaa46c18fabb1888f12c3c0 : SUCCESS in 6m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/29abde9fce9f49d18e87ffb1df7f6b2f : SUCCESS in 1h 24m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/09b740cb257d45729d303b3620df972d : SUCCESS in 1h 54m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/71c90fc517d64d008ee794730b756557 : SUCCESS in 1h 50m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c78a9befd5f492ca07e02d70eb7b30b : SUCCESS in 1h 11m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f3519d2dbea1421e848d2582ce916f0c : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f1bf9b930f12414981921f2936d2fb84 : FAILURE in 46m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e77f4699a9143faa90fb4cf2bd95f03 : SUCCESS in 58m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdc358d9f0484da3bf116f38b05e8439 : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/301ae444a4cb4b9da6aad9eff7a2b14b : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":6},{"id":"551962780c23ffa691ccef03c1fd32ba685e5df3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-20 17:44:31.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b11531b70eb2887b3874e7efe88e25b038782155","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 17:44:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7b0a7dfaa497cd5cf4c111334a289e1470c7a833","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 20:51:41.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/780223b778b64225b3e06b47567fcd49 : SUCCESS in 3m 19s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3a121f6ec6744c19bd798d4fa800cdb0 : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/81ebed6bd5d0472793bc7688037e97db : SUCCESS in 1h 21m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3a8ac67927bf47409b35d18ea0463a7d : SUCCESS in 1h 49m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4d3b349f83a24ee3bebfeafce82d4c49 : SUCCESS in 1h 36m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2aba7f09825482e83f5c23688f01130 : TIMED_OUT in 3h 04m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/da1b5107b39443baa5164d003511143b : SUCCESS in 1h 28m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57ab0054798f4c6ebf190a0f3e6d30db : SUCCESS in 1h 04m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66e5445b3b643b786575cf42a7b9866 : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/922759292a8a40da8461029004dcbcf5 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":6},{"id":"26e437d5dd9c6c2ff585d74449f787ed2e66b421","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-21 10:57:36.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"97d9ebfc2df503d7df1e73fa7d01805471dd58c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 12:50:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a8fc37d6ff324454bc6c4ef8419e1419 : SUCCESS in 4m 39s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/de5b8de9b8ea4c9eab540c65ad3d1857 : SUCCESS in 8m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/fe11586a31de4ed994f16ba0d80d0886 : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/78228939156d46a282e55500163b9490 : SUCCESS in 1h 40m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/02e2bcb8f92d44c6a97bb19a2903471f : FAILURE in 1h 25m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d06ff4296e014cdfa7cc2dcd22292d64 : SUCCESS in 1h 17m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e963e8d25539448a8c86137ff0bfdcda : SUCCESS in 1h 20m 44s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bfcf4f5754174107b1304e90ab6f6813 : FAILURE in 53m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b05a2978fc2c4066bada58229d6dce2d : SUCCESS in 1h 00m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aadcace1a6443fb80a7c31f343ed8d8 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fa7b20983ad4c5fac597409b1639fa0 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":6},{"id":"21b77bf87885afcd809e49c14bc6cabe84fe5bd5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-08-23 02:01:48.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"181b0768cd9b3655e1aff3b7bb798559a267e682","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:07:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/067f1828119e4fa5a07c94d39f01f417 : SUCCESS in 3m 12s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/34efcce4155a4b83ad0e7b0693b05034 : SUCCESS in 7m 26s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f88476b6aee34eb7ac0f62b521bbc6a1 : SUCCESS in 1h 31m 56s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0ad9f77b48694eca99e46cbbf336cf68 : SUCCESS in 1h 43m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/11e5d942731144d0b9728b3b3e564e90 : SUCCESS in 1h 33m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa341ee6677e49658c6a1894f8b35d6e : SUCCESS in 1h 24m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/30ed0300bd794dbbaafa0e39f874f56c : SUCCESS in 1h 20m 00s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/65ff5bc8c238453fa90eaba594829d56 : FAILURE in 54m 09s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7359baae370740a4b6c6327591bf9d12 : SUCCESS in 1h 11m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f351b47e2d34bbba81d4a204355e710 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70130e571bc34b199ee9cc7e06bf0178 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":6},{"id":"189b9b5d17d628c53f556325208b7b50ba4ee120","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 04:08:10.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1c946d29c8e7eb552000e45e8a9ff26e7226b28e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 06:13:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"574cc006a2d276e74fa019c64fdc94ff1ff3c939","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 06:13:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/14335698619a4dfa8268aae407ae8bdb : SUCCESS in 2m 27s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e67b7aa8df834c0b95f0eb3938408118 : SUCCESS in 4m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6eee26654c02413e95f4b4195ff760db : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df1988c3b9774b848d597da1347443f0 : SUCCESS in 2h 03m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b3d2ed93e86b4f09a40252689990babb : SUCCESS in 1h 31m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df8dc8b3b2544a3b8a9ce400046498ee : SUCCESS in 1h 28m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0e5317ed8c7846af9a9aa7c1e719dbd3 : SUCCESS in 1h 18m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf1a4f4e3b454af69a54fd6fa568fd93 : SUCCESS in 55m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2c7f1ae67c48068c5769acfc504c2d : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46f55efdbb8747f0bc1d6321d142ab19 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":6},{"id":"65efed701d8af3ab4be1a33708cbee6cb6799afe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 06:15:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfb3f6d6ec3403ea0f3e472df0fdd77 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6eaddb9daacddf36062919bfac1c251d767950f5","revisions":{"360379eefd226f866f8564a90fec063bbac9dcd5":{"kind":"REWORK","_number":1,"created":"2019-07-18 22:18:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/671599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/99/671599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/1"}}},"commit":{"parents":[{"commit":"eb083ecf9f150cd1d42b31c8bfd17b5038ba004a","subject":"Use systemd-journald instead of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/eb083ecf9f150cd1d42b31c8bfd17b5038ba004a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:18:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:18:18.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2db6ad400d743615a18bcd533228e20d5402970a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/360379eefd226f866f8564a90fec063bbac9dcd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/360379eefd226f866f8564a90fec063bbac9dcd5"}]},"branch":"refs/heads/master"},"b612f8d51060a8309e7be3d70351edc00021bc54":{"kind":"REWORK","_number":2,"created":"2019-07-19 18:21:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/671599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/99/671599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/2"}}},"commit":{"parents":[{"commit":"eb083ecf9f150cd1d42b31c8bfd17b5038ba004a","subject":"Use systemd-journald instead of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/eb083ecf9f150cd1d42b31c8bfd17b5038ba004a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:21:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:21:25.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2db6ad400d743615a18bcd533228e20d5402970a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b612f8d51060a8309e7be3d70351edc00021bc54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b612f8d51060a8309e7be3d70351edc00021bc54"}]},"branch":"refs/heads/master"},"ca3f336efb4006e9acfd163dfc1284f50031b5b6":{"kind":"REWORK","_number":3,"created":"2019-07-20 09:32:19.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/671599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/99/671599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/3"}}},"commit":{"parents":[{"commit":"eb083ecf9f150cd1d42b31c8bfd17b5038ba004a","subject":"Use systemd-journald instead of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/eb083ecf9f150cd1d42b31c8bfd17b5038ba004a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:32:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:32:15.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2db6ad400d743615a18bcd533228e20d5402970a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/ca3f336efb4006e9acfd163dfc1284f50031b5b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/ca3f336efb4006e9acfd163dfc1284f50031b5b6"}]},"branch":"refs/heads/master"},"b24c6939b25ba2f50ef55cda1ea172d0f0698408":{"kind":"REWORK","_number":4,"created":"2019-07-30 19:39:24.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/671599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/99/671599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/4"}}},"commit":{"parents":[{"commit":"57b88ebc65e516322a0268509816ec1b8640b1db","subject":"Migrate gnocchi to use uWSGI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/57b88ebc65e516322a0268509816ec1b8640b1db"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:39:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:39:20.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2db6ad400d743615a18bcd533228e20d5402970a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b24c6939b25ba2f50ef55cda1ea172d0f0698408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b24c6939b25ba2f50ef55cda1ea172d0f0698408"}]},"branch":"refs/heads/master"},"234c8ac0c34180ae98e341ddc6911af1b34cfade":{"kind":"REWORK","_number":5,"created":"2019-08-19 22:20:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/671599/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/99/671599/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/5"}}},"commit":{"parents":[{"commit":"17a2175bdadef30e582ccbb5ee4a09f46f4b7c95","subject":"service_setup: refactor service setup to a single file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/17a2175bdadef30e582ccbb5ee4a09f46f4b7c95"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 22:20:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-19 22:20:47.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2db6ad400d743615a18bcd533228e20d5402970a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/234c8ac0c34180ae98e341ddc6911af1b34cfade"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/234c8ac0c34180ae98e341ddc6911af1b34cfade"}]},"branch":"refs/heads/master"},"6eaddb9daacddf36062919bfac1c251d767950f5":{"kind":"REWORK","_number":6,"created":"2019-08-20 03:06:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/671599/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/99/671599/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/99/671599/6"}}},"commit":{"parents":[{"commit":"17a2175bdadef30e582ccbb5ee4a09f46f4b7c95","subject":"service_setup: refactor service setup to a single file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/17a2175bdadef30e582ccbb5ee4a09f46f4b7c95"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-20 03:05:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-08-20 03:05:59.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2db6ad400d743615a18bcd533228e20d5402970a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6eaddb9daacddf36062919bfac1c251d767950f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6eaddb9daacddf36062919bfac1c251d767950f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
