)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~672706","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I06bbcb2f15108fc517742208ac5291719627ffe2","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"use_uwsgi","hashtags":[],"change_id":"I06bbcb2f15108fc517742208ac5291719627ffe2","subject":"Migrate gnocchi to use uWSGI","status":"MERGED","created":"2019-07-25 12:03:21.000000000","updated":"2019-07-30 08:04:57.000000000","submitted":"2019-07-30 08:04:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672706-1564473897689-8da850ff","meta_rev_id":"f1d3905d61d3fc98633264de27169cf480a322cf","_number":672706,"virtual_id_number":672706,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-30 08:04:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-29 20:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-29 20:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-07-30 05:56:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-29 20:54:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 20:52:03.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-29 20:54:47.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-30 08:04:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb2f41cc3709fb624a72731e7010ffe59050ec4d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-25 12:03:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0763d9117c42b080ad289a0245ecb67c750092b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 13:54:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/672706/1/check/openstack-ansible-linters/2430cef/ : SUCCESS in 6m 46s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/672706/1/check/openstack-ansible-deploy-aio_metal-debian-stable/0a37caf/ : SUCCESS in 1h 14m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/672706/1/check/openstack-ansible-deploy-aio_metal-centos-7/e5e830b/ : SUCCESS in 1h 46m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/06/672706/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/e8fdbee/ : SUCCESS in 1h 38m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/672706/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/bc56e6d/ : SUCCESS in 1h 16m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/672706/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3faba89/ : SUCCESS in 1h 39m 22s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/06/672706/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/76cd798/ : FAILURE in 54m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/06/672706/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/97d91e6/ : FAILURE in 33m 38s\n- openstack-tox-docs http://logs.openstack.org/06/672706/1/check/openstack-tox-docs/96c86f1/html/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"30193835b2834b3db62f9f89eb8dd206d23ce94d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-25 14:06:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6749568f7a41b3f3bf6e8345b5ddf56cee92e2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 16:21:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/672706/2/check/openstack-ansible-linters/854dc8c/ : SUCCESS in 6m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/672706/2/check/openstack-ansible-deploy-aio_metal-debian-stable/604c4df/ : SUCCESS in 1h 37m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/672706/2/check/openstack-ansible-deploy-aio_metal-centos-7/db07495/ : SUCCESS in 2h 08m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/06/672706/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/9daa0b5/ : SUCCESS in 1h 47m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/672706/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/51d4487/ : SUCCESS in 1h 10m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/672706/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/c3fbf36/ : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/06/672706/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/7000f26/ : FAILURE in 52m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/06/672706/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/f9de188/ : FAILURE in 42m 47s\n- openstack-tox-docs http://logs.openstack.org/06/672706/2/check/openstack-tox-docs/c64b5de/html/ : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7e7dabf2f36f495c71b8ca31c7286bce47a945a6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-25 19:47:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f32095bbf18f5964db025c0872d9da1c14418a8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 21:26:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/672706/3/check/openstack-ansible-linters/6e62ed4/ : SUCCESS in 9m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-debian-stable/7d0e452/ : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-centos-7/13dc67f/ : SUCCESS in 1h 31m 28s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/69540b4/ : SUCCESS in 1h 37m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ecc29da/ : SUCCESS in 1h 23m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/3644d23/ : SUCCESS in 1h 24m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/79fae60/ : FAILURE in 1h 01m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/485d2c0/ : FAILURE in 48m 22s\n- openstack-tox-docs http://logs.openstack.org/06/672706/3/check/openstack-tox-docs/4728b9b/html/ : SUCCESS in 3m 42s\n- build-openstack-releasenotes http://logs.openstack.org/06/672706/3/check/build-openstack-releasenotes/6b64165/html/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"27e86322efb54b28877b7d11139531042f7d75ee","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-26 05:10:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ac60660fd4990d3e7d2f5b4e45cf8a5e67879f52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 06:11:23.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/672706/3/check/openstack-ansible-linters/56bf6b1/ : SUCCESS in 4m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-debian-stable/44485b5/ : FAILURE in 46m 14s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-centos-7/6573543/ : FAILURE in 53m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/b85aae7/ : FAILURE in 53m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3cd58fe/ : FAILURE in 41m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/0cf4e7f/ : FAILURE in 46m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/329cd97/ : FAILURE in 41m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/38d5b82/ : FAILURE in 32m 08s\n- openstack-tox-docs http://logs.openstack.org/06/672706/3/check/openstack-tox-docs/18ff190/html/ : SUCCESS in 2m 35s\n- build-openstack-releasenotes http://logs.openstack.org/06/672706/3/check/build-openstack-releasenotes/fb01d04/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":3},{"id":"12cab58745875284b1fd73693b390290044acd92","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-26 19:49:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6bea98379affc3e3e50a2dace1295a1fc4323011","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 21:12:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/06/672706/3/check/openstack-ansible-linters/935502c/ : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-debian-stable/1b55e87/ : FAILURE in 49m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-centos-7/160c5ac/ : FAILURE in 1h 01m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/52abd0d/ : FAILURE in 54m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5043869/ : FAILURE in 48m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e7064b5/ : FAILURE in 55m 47s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/40f9928/ : FAILURE in 50m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/e513ab5/ : FAILURE in 40m 55s\n- openstack-tox-docs http://logs.openstack.org/06/672706/3/check/openstack-tox-docs/6679b71/html/ : SUCCESS in 4m 02s\n- build-openstack-releasenotes http://logs.openstack.org/06/672706/3/check/build-openstack-releasenotes/9a2385d/html/ : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":3},{"id":"ee2c5cd685a3820ec7abf5cf9d6e9e05d0e758ce","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-27 08:24:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"5dccfa83c5c37c9329e83dfec7749dfcf6bd5239","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-27 10:06:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/06/672706/3/check/openstack-ansible-linters/67abf62/ : SUCCESS in 5m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-debian-stable/c3539e7/ : SUCCESS in 1h 17m 24s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-centos-7/08a08f1/ : SUCCESS in 1h 36m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/3db7b8c/ : SUCCESS in 1h 32m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/65b09bf/ : SUCCESS in 1h 41m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/c63e39b/ : SUCCESS in 1h 09m 41s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/0f14fb2/ : FAILURE in 1h 06m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/4a825d3/ : SUCCESS in 53m 06s\n- openstack-tox-docs http://logs.openstack.org/06/672706/3/check/openstack-tox-docs/ce09a4e/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/06/672706/3/check/build-openstack-releasenotes/1708a86/html/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":3},{"id":"8d62d37290ab08e4c1d95d8d4b2a0b8c1beecd97","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-29 20:52:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a3cd9a84caf0065cab3e00983b0495a7be6779dc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-29 20:54:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d20e3358e2183b48055296e14669b9717134ffd8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-30 05:56:30.000000000","message":"Patch Set 3: Workflow+1\n\npush to the gate","accounts_in_message":[],"_revision_number":3},{"id":"6972531b265d385f5f298e44f3c3bed7d71493f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 05:56:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"00e040be2d4bc9de4bec0e5dd6780bfcd27215a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 08:04:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"eb3041307e9130ea619328d2978e9a81e0b074ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 08:04:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/06/672706/3/gate/openstack-ansible-linters/2f7d76c/ : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/06/672706/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/5799d2d/ : SUCCESS in 1h 39m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/06/672706/3/gate/openstack-ansible-deploy-aio_metal-centos-7/29f28e7/ : SUCCESS in 2h 02m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/06/672706/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/8588650/ : SUCCESS in 1h 40m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/021d2fa/ : SUCCESS in 1h 22m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/06/672706/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/a73662f/ : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/06/672706/3/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/834b6aa/ : SUCCESS in 1h 05m 39s\n- openstack-tox-docs http://logs.openstack.org/06/672706/3/gate/openstack-tox-docs/1b4da95/html/ : SUCCESS in 2m 46s\n- build-openstack-releasenotes http://logs.openstack.org/06/672706/3/gate/build-openstack-releasenotes/b768009/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"57b88ebc65e516322a0268509816ec1b8640b1db","revisions":{"f4e68f14909be9de84ffe955fb450fce3a2c2acb":{"kind":"REWORK","_number":1,"created":"2019-07-25 12:03:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/06/672706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/06/672706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/1"}}},"commit":{"parents":[{"commit":"eb083ecf9f150cd1d42b31c8bfd17b5038ba004a","subject":"Use systemd-journald instead of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/eb083ecf9f150cd1d42b31c8bfd17b5038ba004a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-25 12:03:04.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-25 12:03:04.000000000","tz":180},"subject":"Fix gnocchi wsgi scrpt","message":"Fix gnocchi wsgi scrpt\n\nGnocchi wsgi launch method has changed a bit, so we\u0027re bringing it\ninto accordance with gnocchi-api binary\n\nChange-Id: I06bbcb2f15108fc517742208ac5291719627ffe2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f4e68f14909be9de84ffe955fb450fce3a2c2acb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f4e68f14909be9de84ffe955fb450fce3a2c2acb"}]},"branch":"refs/heads/master"},"61db936eb75f3eff9ba73e025092cca70eb6d41c":{"kind":"REWORK","_number":2,"created":"2019-07-25 14:06:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/06/672706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/06/672706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/2"}}},"commit":{"parents":[{"commit":"eb083ecf9f150cd1d42b31c8bfd17b5038ba004a","subject":"Use systemd-journald instead of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/eb083ecf9f150cd1d42b31c8bfd17b5038ba004a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-25 12:03:04.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-25 14:06:03.000000000","tz":180},"subject":"Fix gnocchi launch","message":"Fix gnocchi launch\n\nGnocchi wsgi launch method has changed a bit, so we\u0027re bringing it\ninto accordance with gnocchi-api binary\n\nSince gnocchi is using apache, it\u0027s logs should be forwarded as well.\nOtherwise apache will fail to start with undefined log directory.\n\nChange-Id: I06bbcb2f15108fc517742208ac5291719627ffe2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/61db936eb75f3eff9ba73e025092cca70eb6d41c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/61db936eb75f3eff9ba73e025092cca70eb6d41c"}]},"branch":"refs/heads/master"},"57b88ebc65e516322a0268509816ec1b8640b1db":{"kind":"REWORK","_number":3,"created":"2019-07-25 19:47:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/06/672706/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/06/672706/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/06/672706/3"}}},"commit":{"parents":[{"commit":"eb083ecf9f150cd1d42b31c8bfd17b5038ba004a","subject":"Use systemd-journald instead of log files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/eb083ecf9f150cd1d42b31c8bfd17b5038ba004a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-25 19:47:29.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-25 19:47:29.000000000","tz":180},"subject":"Migrate gnocchi to use uWSGI","message":"Migrate gnocchi to use uWSGI\n\nThis patch moves gnocchi-api from usage of apache with mod_wsgi\nto uWSGI, which means unification across another roles and\nreduced maintenance costs\n\nDuring migration period tasks that ensures apache won\u0027t listen\non gnocchi_service_port are present, but they are supposed to be removed\nafter train release.\n\nDepends-On: https://review.opendev.org/671988\nChange-Id: I06bbcb2f15108fc517742208ac5291719627ffe2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/57b88ebc65e516322a0268509816ec1b8640b1db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/57b88ebc65e516322a0268509816ec1b8640b1db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
