)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~679991","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Iec03bd79279e694678336880460bcb83f68d9780","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"ansible-role-uwsgi","hashtags":[],"change_id":"Iec03bd79279e694678336880460bcb83f68d9780","subject":"Start using uWSGI role","status":"MERGED","created":"2019-09-04 13:26:02.000000000","updated":"2019-09-07 01:57:20.000000000","submitted":"2019-09-07 01:56:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679991-1567821376048-72f0986f","meta_rev_id":"f5c4df9ddeb50297fbcd2732308883b90d05b341","_number":679991,"virtual_id_number":679991,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"date":"2019-09-06 21:33:42.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-07 01:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-09-06 02:06:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-06 13:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-06 13:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 02:06:45.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-09-06 13:53:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-07 01:56:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f515046318b3f2632e67a525ff6aad7a5e35f49c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-04 13:26:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37846b7d8ecb330195dbc692360a915ce6854711","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-04 14:36:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ce7946dc32e05ad658a5e4a31fd928c5dc50b4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 17:05:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ef808831a2834fd5801bc86a4b4eaa5b : SUCCESS in 7m 47s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ddb5202014ac41c282b9086775dee590 : SUCCESS in 1h 25m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0f14a19331f4488dbe5d5711b401a356 : SUCCESS in 2h 25m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4b8684c978104de381aaac51eed6ccd4 : RETRY_LIMIT in 6m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56332fc8bfdd4070ac43cc178dd915da : FAILURE in 1h 31m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/35b7c728a3794d359819c60b4e45a291 : SUCCESS in 1h 22m 31s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d9ecd87ba11d46eabf9642efe9b0ae3c : RETRY_LIMIT in 6m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab2dec2a287448ca8e0e869714bd2956 : SUCCESS in 1h 09m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff87042dc324f7393b6773c8579cad5 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1dda27eea1d1880c27564d590115860eaa512439","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 15:58:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4e278ae3502b23953d180139cf7866828b84c282","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 19:21:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6eea434e5a844ecaad8be572a7d9b1b3 : SUCCESS in 6m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/760864f3e8f94c16b850e6a6548cd668 : SUCCESS in 1h 26m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e288e6a259eb4bcc80205d0cbcb43b66 : SUCCESS in 1h 50m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b5381ff735314d28a5d851d9f0e806bc : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15a50f5485ba4336ae6d015aa4078539 : SUCCESS in 1h 30m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c4cca3b8a15e4b87af401310e5d010a9 : SUCCESS in 1h 27m 08s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/99cf7b1d5dc14b2b9ad17b3f0ce5d979 : FAILURE in 32m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e81122c5bce418eb9b2061345886692 : SUCCESS in 1h 01m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f74538c8ae9e422a964db9e27c12b90f : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"08dca1039f692475a412bbe773b3ed44d42acb4f","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-09-06 02:06:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d7dde928fe5626a28ebf928bf0a014d151d9a3a6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-06 13:53:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6db3157b2347bc874b9e4836a4471e5674582060","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 13:53:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09d25043acac1e3db0bb712741c71b04401b3018","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 15:53:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3aa53946cfc4464096f00c3e077e42d1 : SUCCESS in 8m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2cc5bec2ace34a58b3727a373ef2062f : SUCCESS in 1h 43m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8f785257d0d748689a3cdfa36207361e : SUCCESS in 1h 49m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e59c15df96f849ca8f5ee7eeba4a0b04 : FAILURE in 1h 33m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5b38784a9a54e0bb66164888e92877b : FAILURE in 1h 19m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/641a6bb0513b44c181c82b48bc50318f : SUCCESS in 1h 18m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aebd674507e044a68e47127ea0f9f497 : SUCCESS in 1h 02m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85bcd860c5d74e14968cbd381772c22d : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0d6cc5f5bf3cdc0f9c927a82c6aa1a1dcb83bb16","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 16:28:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"116d06973d33bc7a550ac2001f6741c09fa99900","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:52:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e9505a911eb441bfa35d56e3696d6ea2 : SUCCESS in 9m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/a5f15f48bc654d6e9feba42560de7e58 : SUCCESS in 1h 28m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aedcea230542499a9d8598df4a037a3c : SUCCESS in 2h 06m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/28ef387f1f89486e865c992afc10d9ec : SUCCESS in 1h 52m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1c29cc2fbd44b48a0cc59aeab17d729 : SUCCESS in 1h 30m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e04411de6c3b49f5843eed8693578bfb : SUCCESS in 1h 46m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/538cd1ee6b76417fafb404585762faef : FAILURE in 29m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/077318f62b0d44beab76c3bc88632029 : SUCCESS in 1h 02m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4c92f98cd74e089bd6080622349a70 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":2},{"id":"ed75bb03e07a90f068974d0e15c3a9f0bc8d7525","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:52:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d86d4e2354148391ddd50948f5050c68c4a51b21","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 21:00:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/26588c14ba41447ba563e45b80f3b996 : SUCCESS in 7m 10s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/16ff60bd310141458e221f353d387f87 : SUCCESS in 1h 54m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/911c5f2e968243cdafd366c6ca1f52e5 : SUCCESS in 2h 00m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9f70169db35d447d9e05d3c05624539f : SUCCESS in 1h 35m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c253d3381efc46e99a04dba232abf393 : SUCCESS in 1h 25m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/46a08c0e60bd40feabe0c449a0678416 : SUCCESS in 1h 26m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07440e782f964750a7eebd7010c4376f : SUCCESS in 1h 04m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9fb8bff8253417e92b6c7eab7db57cb : POST_FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"404a6f296456caf1c9b6a0c58d888872780a7798","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 21:33:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"44d07a1b14a51449f4ad0229508e51c71c0f10b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 23:32:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/97e3b17c8dce496ca68c856302b40e00 : SUCCESS in 5m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/070cecc74a484a44adc98ad9e0820868 : SUCCESS in 1h 23m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eca9bcfeab6f4800b0b44957f38bcbab : SUCCESS in 1h 49m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9e4d00f517c9473da9f3c55446dc50a8 : SUCCESS in 1h 40m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89de1a141f35496dad6c8456301ffcc4 : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c11fe63cc2f8497c91646d18537f27fb : SUCCESS in 1h 34m 59s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ed946a719172470992f2c9e36f30ad82 : FAILURE in 32m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69fb0c3442994418bdf203f5eb3da003 : SUCCESS in 1h 01m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e8b1e156434b97932e9b23a843c5f6 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":2},{"id":"316613a3c7fb4715e63e958a390a57c2da3829a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 23:33:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"117674710273433a0f595dd3d6aa7cb225b1310b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 01:56:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e5cc62a5d3b94b76b1144d9a2a1fb1a7 : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f946281f7ab346759a8b79effbbf3cc7 : SUCCESS in 1h 28m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6a180ba798144e30bb578044195012ec : SUCCESS in 2h 17m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b3d9b9bb98f14a07ad85f4eb55a57642 : SUCCESS in 1h 44m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22eb458d7a524ce09a043cdab9169867 : SUCCESS in 1h 35m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5574a9e65461467aa1d9421bd216de2b : SUCCESS in 1h 44m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a970c79043214ecd9bf9a209c8682354 : SUCCESS in 1h 06m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e363554daadd41c0af45b36394987a04 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0276ba584b5de8af82e2e07d726fb3842d30b8ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 01:56:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"abfa8a2d58b9a1299b74ae045bfee8cece86decd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-07 01:57:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc83663affc94db9a5eb0cf9106b89c2 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b7a76fb3412f0b499afb0625ca5613801dabeecc","revisions":{"3f9d160965bfb9a7ae723cb2391129e5e227c208":{"kind":"REWORK","_number":1,"created":"2019-09-04 13:26:02.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/679991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/91/679991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/91/679991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/91/679991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/91/679991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/91/679991/1"}}},"commit":{"parents":[{"commit":"6eaddb9daacddf36062919bfac1c251d767950f5","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6eaddb9daacddf36062919bfac1c251d767950f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-04 13:25:51.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-04 13:25:51.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: Iec03bd79279e694678336880460bcb83f68d9780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/3f9d160965bfb9a7ae723cb2391129e5e227c208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/3f9d160965bfb9a7ae723cb2391129e5e227c208"}]},"branch":"refs/heads/master"},"b7a76fb3412f0b499afb0625ca5613801dabeecc":{"kind":"REWORK","_number":2,"created":"2019-09-04 14:36:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/91/679991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/91/679991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/91/679991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/91/679991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/91/679991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/91/679991/2"}}},"commit":{"parents":[{"commit":"6eaddb9daacddf36062919bfac1c251d767950f5","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6eaddb9daacddf36062919bfac1c251d767950f5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-04 13:25:51.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-04 14:36:30.000000000","tz":180},"subject":"Start using uWSGI role","message":"Start using uWSGI role\n\nMove service to use uWSGI role instead of iternal task for uwsgi\ndeployment. This aims to ease the maintenance of uWSGI and speedup\nmetal deployments as the same uwsgi environment will be used\nacross all services.\n\nChange-Id: Iec03bd79279e694678336880460bcb83f68d9780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b7a76fb3412f0b499afb0625ca5613801dabeecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b7a76fb3412f0b499afb0625ca5613801dabeecc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
