)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~680044","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I7ffbf584205b27d9e010cbba3b24902c866a0e37","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I7ffbf584205b27d9e010cbba3b24902c866a0e37","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-09-04 15:14:02.000000000","updated":"2019-09-06 20:21:36.000000000","submitted":"2019-09-06 20:20:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680044-1567801224740-0cd57bd2","meta_rev_id":"ce38ab1dfa9a0516baea81638ab251a8da6a10d8","_number":680044,"virtual_id_number":680044,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-06 20:20:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-06 09:51:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-09-06 09:51:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 09:51:40.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-06 20:20:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9807fe7e7fd09dbbbe0652e791c4fa1b3de6da02","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 15:14:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4efb45f51fbd0882cb568e50541004483cfcfaa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 18:07:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c8a520da069741538020dbce043a500d : SUCCESS in 6m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9e3807b3cc9040f6b8bd92100e4c3797 : POST_FAILURE in 1h 16m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/497042ed241b45008db428b155c5a062 : POST_FAILURE in 1h 46m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/60be3df69f534857b09389c7224c1f79 : FAILURE in 37m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13a3f1ba921c4cd7a561800cb62060e5 : SUCCESS in 1h 11m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5023a8b225b84d8d82acfc202878d713 : SUCCESS in 1h 39m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/43a216a12e7c4a1aab1c760d065776ba : FAILURE in 45m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f40cde54f1f043a68ca1d38b6fa05bc2 : SUCCESS in 1h 08m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/063fff804e5444d886a8fa19f60c49f6 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7f4a1b4a7801c4ec631dafa1f2ddad0480f9fee5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 14:45:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2a3df2065bd01133134470c722d12de3629e28b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 18:24:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/220edf05183d46a8b2be4d54a87c42ec : SUCCESS in 9m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b4a7b9609d0a4436be41e2878c1ea820 : SUCCESS in 1h 44m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/275f33dbe6b1485a85ad9c7f92a2bac3 : SUCCESS in 1h 54m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ee87a278f960452b84b0053231db6caa : FAILURE in 1h 52m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af906f73902f4c5380f9219684a40650 : FAILURE in 1h 37m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f6539f2b4b234c7f9e27a7dba70ddc23 : SUCCESS in 1h 59m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4535e623c9a94a0ca009cea45eeaf398 : FAILURE in 42m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/075e1c97e143421588b8dae7340d2c56 : SUCCESS in 1h 16m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf68f9b5dfc428ea62bf09244b765a9 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7f8b4cbb9cbc9c979e878fb1c3141898c7183f0d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 21:54:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c4f7d24e5f1314f664b3ebb3cf785c064a71b4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 00:34:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f7203eaec704431c873f1ff42076fd78 : SUCCESS in 9m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/727bb66d3b234606a98eb99690c9e7ad : SUCCESS in 1h 21m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ba31e403009a42c2a5e9442495b595d3 : SUCCESS in 2h 11m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/13f7fdc078ed4af683a866a66edeec63 : FAILURE in 2h 30m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2b573f29faa403eab3c6a7477f3df5d : SUCCESS in 1h 36m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d29ad0772dd64970a436183f52f26c8a : SUCCESS in 1h 41m 25s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ddaa0f8eb4254b1c8bfc9aabba009ff2 : FAILURE in 34m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/416728f9dbf74efa8d60b44d41687921 : FAILURE in 57m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c22d1a82c94c59a423f010032a09ba : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c34edea396680177ecd879ef15230fc0a9c55f73","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 07:10:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd2690328704c4f3f67fcf50afdeb5e0055d39c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 09:21:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ad1c52690aff4c0bbfde7ebf1e206b4e : SUCCESS in 6m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/556be4b3ff884ab8a1344792c6414327 : SUCCESS in 1h 27m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/51c4b347f38649a691e0849aec5b39fe : SUCCESS in 1h 46m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3ada28689e4e4d1c984da094e3fd80a4 : SUCCESS in 1h 39m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dabe4a7183594b0a96c9507edab06b1a : SUCCESS in 1h 35m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/54362f4400944d4ea1c87dd52b162177 : SUCCESS in 1h 26m 11s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8f547d7f744e452eb498e099c6a339a2 : FAILURE in 30m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f16a823a22b34a679f62c51d6a3be46b : SUCCESS in 54m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c17f8678ef04357a8d49bf44d3dd841 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8b56659d94bf3743dd84956673b11ab621f68053","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 09:51:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82c1d259dd5689c7e0ea323bec83e8edc21ddbaf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 09:51:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ca4e189e8f067df9f74c76ea1ad5c7ffcff9cca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 11:42:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f68082f08b3f4894b563446ad56c8090 : SUCCESS in 7m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ec0beef2dc7e4bb7863837ddc5b134cb : SUCCESS in 1h 20m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e7a84e2220b24749b1dd5f4d6686a01c : SUCCESS in 1h 44m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f6ca8b19fac345d68857157d6fe45bb9 : FAILURE in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/158f207eb44146d0b5060d36ee8ec1d3 : SUCCESS in 1h 12m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/65dc3204e56846fa89222c16ae70e40b : SUCCESS in 1h 12m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d591e4cb10804597879cb159a8642aad : SUCCESS in 1h 19m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0e7282cb1547c9b0b388cd519dfba7 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"87bc24a0c4edd47a213b04795cdc4f15460868e6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 16:16:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eaa8e919da4bdb0d9892019bf0ae42c28b20c295","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 16:29:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9150668295fc3d95f8196855d82e8166df25b8fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:27:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5b5f87a9efe44dbb97f59b0ef5f8211a : SUCCESS in 5m 47s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3069832220294f2099b30bc9413f4049 : SUCCESS in 1h 50m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c423597a81664b5098fc17db1b36a10e : SUCCESS in 1h 52m 53s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a2f20652776b40e39483dc3c16f3dd4a : SUCCESS in 1h 48m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da004d428b0b4d43bf012ebbae9cf915 : SUCCESS in 1h 31m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e68dfa409fdb4bec8d065f9432b95c52 : SUCCESS in 1h 43m 15s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5ef4cda0a7474cb5a055a7ff68cad376 : FAILURE in 28m 47s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c2bc24145a1486a9b27a89b5113c0dd : SUCCESS in 1h 18m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ebfe8d676564e76a876f2b182c3253a : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":1},{"id":"3e28d14ef388579bd051bd4958aa9977b4651707","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 18:27:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af20145584d2eadb34b31524c987e98b6e38718f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 20:20:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e227b9806f1ec2976f9af5fcf33fad4376109599","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 20:20:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fab73148c1df44209bf669ed305be742 : SUCCESS in 6m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e81c4db7947442e88e2cac176bb23d4c : SUCCESS in 1h 21m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/94a22e2c65c34f01bb585425b9ebf01a : SUCCESS in 1h 49m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b3c73e898df64ecebb283051cde57033 : SUCCESS in 1h 35m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/539fb2504ce6484aa791533cd99f9960 : SUCCESS in 1h 40m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3e6243ddb715439290f6b994bbbcd9b6 : SUCCESS in 1h 34m 50s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec6c21b654f8429babd74e90b875d580 : SUCCESS in 1h 18m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734f2cb820924221814d7680f667abe7 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a0c0544bc5f3d922d0c12ac1560bf6f6c147e64a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 20:21:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02ed5f9667034c8c8e9e4c0349def74c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4b64f38a446537abada78b1cca03077de7927ce","revisions":{"a4b64f38a446537abada78b1cca03077de7927ce":{"kind":"REWORK","_number":1,"created":"2019-09-04 15:14:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/680044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/44/680044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/680044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/680044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/680044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/44/680044/1"}}},"commit":{"parents":[{"commit":"6eaddb9daacddf36062919bfac1c251d767950f5","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6eaddb9daacddf36062919bfac1c251d767950f5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 15:13:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 15:13:58.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I7ffbf584205b27d9e010cbba3b24902c866a0e37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/a4b64f38a446537abada78b1cca03077de7927ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/a4b64f38a446537abada78b1cca03077de7927ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
