)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~688477","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~stable%2Fqueens~I5ebd65c4a0bcda691d5c099b19fa6f66f7025e0f","project":"openstack/openstack-ansible-os_gnocchi","branch":"stable/queens","topic":"goodbye_42.3","hashtags":[],"change_id":"I5ebd65c4a0bcda691d5c099b19fa6f66f7025e0f","subject":"Remove openSUSE Leap 42.3  jobs","status":"MERGED","created":"2019-10-14 17:06:29.000000000","updated":"2019-10-14 19:08:06.000000000","submitted":"2019-10-14 19:08:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688477-1571080086329-14372de2","meta_rev_id":"5709daee185c8b89416e39a262a9771f7275c2d8","_number":688477,"virtual_id_number":688477,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 19:08:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-10-14 17:30:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-10-14 17:30:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 17:30:01.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-14 19:08:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a55aa94d55313bebaca068e9c6bd91a77931b5a3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-14 17:06:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff88ca6d8ae8e84bfc6e9993a4a009dca68b7f05","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-14 17:30:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2bd42638a9c11e21db34ad2200d156d2891d65ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:07:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e78e0862ffd64ba3834215f25e890f86 : SUCCESS in 3m 55s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/365697960dd943bda02d95afd34ea301 : SUCCESS in 7m 00s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/8d24fcc848be44deb24c56b92d73d017 : SUCCESS in 53m 00s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/fb89dab91d924548b47650c6c7c0c269 : SUCCESS in 26m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8311fddecae3b2d677c25e8668f05dd14fce9db8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:12:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c863e8122bac43cc67a47997acc27469fedb88a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:08:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"04ed7c3ab74366101ccfdb1f42db5756241be688","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:08:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/40ad43733b2a400d94e44fddb6886671 : SUCCESS in 2m 48s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5939f203406e4002a1c129c3deb2030a : SUCCESS in 6m 13s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/583404943f594dd28e99d7c85e8f6751 : SUCCESS in 52m 18s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/b43f763a17714d15b3936e69768ef7d3 : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef60f71bf769fcf47793360b0ffd4dada657988f","revisions":{"ef60f71bf769fcf47793360b0ffd4dada657988f":{"kind":"REWORK","_number":1,"created":"2019-10-14 17:06:29.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/77/688477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/77/688477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/688477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/688477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/688477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/688477/1"}}},"commit":{"parents":[{"commit":"51b558103ef9c90b19d1aea455ca3a121eb5dcf8","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/51b558103ef9c90b19d1aea455ca3a121eb5dcf8"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 17:06:15.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 17:06:25.000000000","tz":120},"subject":"Remove openSUSE Leap 42.3  jobs","message":"Remove openSUSE Leap 42.3  jobs\n\nopenSUSE Leap 42.3 is EOL since June 2019, let\u0027s remove remaining jobs so\nthat we can remove Leap 42.3 from Opendev infrastructure.\n\nChange-Id: I5ebd65c4a0bcda691d5c099b19fa6f66f7025e0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/ef60f71bf769fcf47793360b0ffd4dada657988f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/ef60f71bf769fcf47793360b0ffd4dada657988f"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
