)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~690247","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~stable%2Ftrain~Ib93027f6b8bdc2827a7fee127c1f12dc2f0489cb","project":"openstack/openstack-ansible-os_gnocchi","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"Ib93027f6b8bdc2827a7fee127c1f12dc2f0489cb","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","status":"MERGED","created":"2019-10-22 18:58:22.000000000","updated":"2019-10-24 14:17:44.000000000","submitted":"2019-10-24 14:15:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690247-1571926525674-0fd08e9d","meta_rev_id":"ebda14c7a90a13cdb85f13e828d0b9c76aab7b3f","_number":690247,"virtual_id_number":690247,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 14:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-10-24 14:02:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-10-24 14:02:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 14:02:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-24 14:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ba5dc2a7130c52231a1ba4505ec5c45f3d4c423","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-22 18:58:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbed668f178061fd9538462d94e367f899898aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:25:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/489a3c10549b4e048ec5e8b159e8daa8 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31ee0925ed848efac8b212a53f08c7a : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa5037e265c046df93904760fa3b4f62 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"406ef3147f4aecf7a20975d2f3ce36d1c7daaaa4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-10-24 14:02:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"322b09bf049b84542fbfe7acd13b28b0e5df3f2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 14:02:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b85c4b111641e5bb2c53a90409b23176574eed6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 14:15:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/315f9de8c7f246dd86d72df37a89b09f : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7bb96fb9e274be384d885b4a34a9d3b : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cbaf582b8e04d5985123fcd7bb601c0 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"bb2de661cad08b11a8ed271799ed3fd6e47d2237","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 14:15:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ba6b9cf0cc3d871568f453794ded8cba7765809b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 14:17:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93580b592ce44bceb64c8d28fb8cd725 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0b271b6034e4f4dbce30658ae873469 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e4109ea28e97a926b6c8c2679aae77893cc5209","revisions":{"9e4109ea28e97a926b6c8c2679aae77893cc5209":{"kind":"REWORK","_number":1,"created":"2019-10-22 18:58:22.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/47/690247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/47/690247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/47/690247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/47/690247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/47/690247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/47/690247/1"}}},"commit":{"parents":[{"commit":"6bd79209b9e5ebfdce83f4425267f45831b8003c","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6bd79209b9e5ebfdce83f4425267f45831b8003c"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-22 18:58:20.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-22 18:58:20.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/train branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ib93027f6b8bdc2827a7fee127c1f12dc2f0489cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/9e4109ea28e97a926b6c8c2679aae77893cc5209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/9e4109ea28e97a926b6c8c2679aae77893cc5209"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
