)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~718626","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Ib6ccf4102ffc2fc659554e31e480b22b1bfa72d9","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa-collections","hashtags":[],"change_id":"Ib6ccf4102ffc2fc659554e31e480b22b1bfa72d9","subject":"Use ansible openstack collection","status":"MERGED","created":"2020-04-09 09:22:50.000000000","updated":"2020-06-08 20:56:06.000000000","submitted":"2020-06-08 20:29:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718626-1591648143691-237d3608","meta_rev_id":"8d6f8b18408073d8e4fe5e7c33ed3ffd4217a525","_number":718626,"virtual_id_number":718626,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 20:29:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-06-08 19:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-06-03 10:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2020-06-08 19:05:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 10:28:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-08 19:05:34.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-08 20:29:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6949b291e8be1036953b33bd617f47f2f7f695fc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 09:22:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d70440b74b0d1cf85d7584cbac9393d3e86d66b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 09:54:18.000000000","message":"Topic set to osa-collections","accounts_in_message":[],"_revision_number":1},{"id":"2faeb17745c9e9a4b2f7bbe470e277a5b1052808","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 11:54:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b66fbe933dc440f6ba7a10527c4d9595 : SUCCESS in 9m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/355ba65909054f5cbdee456aabe3d305 : SUCCESS in 1h 24m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/607c14156b1947298f33bbb297a2363c : SUCCESS in 1h 43m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/372c4cb75b984f1e87472344bd60c014 : FAILURE in 1h 25m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40c03ac971114b4484546342a1f2b2bc : SUCCESS in 1h 48m 52s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/29d3ebc0e8564b7da0ffdb407149d264 : FAILURE in 1h 51m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0579c1dd16642cca9fbca7d65a4e1bf : FAILURE in 1h 29m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a902b8b8e1334014aacb5658d4e41139 : SUCCESS in 1h 24m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/789f7f1154894131935590e66e89bca0 : FAILURE in 51m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53edeadd530b4fc1afe11a2865bb3a23 : SUCCESS in 1h 20m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c95652f43e54565b9e69a8d415b89c3 : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3624c2e66035aecd745c81785a9bf0125418bfdd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 15:07:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6608efafe65cb8942a79b93a4b0648b6ea4ed4d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 18:05:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6d5aecd01514476e84ea7c79c673f508 : SUCCESS in 1h 20m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6c427cf5ea644ba29ce3f628ce203235 : SUCCESS in 2h 06m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f15007a0f7a94dfcb4170b8dc1949570 : SUCCESS in 1h 48m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bead994a2e5467c872aa3141953ed71 : SUCCESS in 1h 35m 37s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ea9989788db142b2b86778baf6cc527c : FAILURE in 2h 01m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32cfbf55ccef4e1782404a31076e8231 : FAILURE in 1h 25m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/57213b4a4c6b4105b2e611d1fb6a869f : SUCCESS in 1h 38m 40s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/74df2a6662d9409599a12f7b71759e71 : FAILURE in 38m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccb15459da9c4b3fac44c760f4ab8768 : FAILURE in 31m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95810216be474d0a99bb1abd8ac12b27 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f012cece055cf6375d425c85bdbac3dd54e5366b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 12:39:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d490cda142a7a4df4c53b8425ee3c479f0ac66d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 15:24:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8a7934a461e345288018b5f11c2b2c34 : SUCCESS in 11m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9579b384fd344e9aa4e89e5f0c02f3eb : SUCCESS in 1h 25m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9bc680894e1543978a62aa41a0918886 : SUCCESS in 2h 04m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3532e1c011184af2a29e84ee3fb04c2c : SUCCESS in 1h 34m 09s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0f8597a674164af885d6e24830542dea : FAILURE in 1h 26m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22f240cb599b430c9b75fa9f02d50e80 : SUCCESS in 2h 36m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8dbef46fc30f4061a0749169db305465 : SUCCESS in 1h 25m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/68b22b4028e74150b9379e53b266ee4e : SUCCESS in 1h 09m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a54e4d2b24490e967774dd86d7dfad : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":2},{"id":"aee08b5bc30ba4173a2e34ac2cda0ee9841b4cda","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-03 10:28:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"49e865785e88be9d7f5e0c557a13021561733870","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-08 19:05:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d446f25a409fc14e2e8728ccbd2d99ffeaa5bbdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 19:05:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bba6a269abde6bc958aab79d8c86b5c8c5700274","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:29:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a78b8f49f8d5d1039a8000dd47ad115f45516089","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:29:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/779ed4c855534cc9b6e1b936c30e80a3 : SUCCESS in 8m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0ffdad5e6d224add82fcc5eaea27395b : SUCCESS in 1h 00m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c72b65cbc540483dbcb7aae92fb2c524 : SUCCESS in 1h 14m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bd11ff0ab1441e997326ea75f55a930 : SUCCESS in 1h 06m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0b3b7b78b4534653ad806662a93ca857 : SUCCESS in 1h 05m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08a71deaed904e678aca38d0b292a7dc : SUCCESS in 53m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c6e7fa915374a58bd527558cae5cf70 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b896a603d01f2434cfd6a6e65ffa7750748d65be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:32:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d719a847db9c4122824c13e8e1807536 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8d6f8b18408073d8e4fe5e7c33ed3ffd4217a525","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 20:56:06.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 60e5ddb4f6022450761049b918779e1350d2af2c","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4067c265420bc27062e21616323a9c269ea7a86b","revisions":{"c003e8cbf9bbc80becc6e3c618f0b0a8624b68e0":{"kind":"REWORK","_number":1,"created":"2020-04-09 09:22:50.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/26/718626/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/26/718626/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/26/718626/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/26/718626/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/26/718626/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/26/718626/1"}}},"commit":{"parents":[{"commit":"0490611056700c7e57cf717d9ed0a712875206fb","subject":"Refactor memcached_servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/0490611056700c7e57cf717d9ed0a712875206fb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:22:04.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:22:46.000000000","tz":60},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ib6ccf4102ffc2fc659554e31e480b22b1bfa72d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/c003e8cbf9bbc80becc6e3c618f0b0a8624b68e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/c003e8cbf9bbc80becc6e3c618f0b0a8624b68e0"}]},"branch":"refs/heads/master"},"4067c265420bc27062e21616323a9c269ea7a86b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-20 12:39:50.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/26/718626/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/26/718626/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/26/718626/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/26/718626/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/26/718626/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/26/718626/2"}}},"commit":{"parents":[{"commit":"1f94763fad23a500b3a9e530f13a6dc142a57baa","subject":"Merge \"[ussuri][goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1f94763fad23a500b3a9e530f13a6dc142a57baa"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:22:04.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-20 12:39:50.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ib6ccf4102ffc2fc659554e31e480b22b1bfa72d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4067c265420bc27062e21616323a9c269ea7a86b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4067c265420bc27062e21616323a9c269ea7a86b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
