)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~725154","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I7b88a053ac5d779bc53a32c1294e76e2233cdd9e","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","hashtags":[],"change_id":"I7b88a053ac5d779bc53a32c1294e76e2233cdd9e","subject":"Fix rados linking for py3","status":"MERGED","created":"2020-05-04 03:07:05.000000000","updated":"2020-05-06 18:46:08.000000000","submitted":"2020-05-06 18:38:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725154-1588790323209-a9cd72ef","meta_rev_id":"482c01cd67967ded6b5ef3676295274d25dd3b68","_number":725154,"virtual_id_number":725154,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-06 09:09:38.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-06 18:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2020-05-06 16:42:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-06 12:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2020-05-06 16:42:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-06 12:28:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-06 16:42:57.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-05-06 18:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7999e5efc3cd922ed01e293770f77c63c3ba7fb6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-04 03:07:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ad9b45615f07ef23344e49a019b9203ff577cd3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-04 03:46:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"520644db12fa59b605dc3e6b0f2efd912cab51e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 06:06:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fda319a1cc42414981e869873728ed68 : SUCCESS in 6m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f8751b2def504185900c30defa637466 : SUCCESS in 1h 27m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0e7aae6e6b9e46ef9ac41cdb12b61c7f : FAILURE in 16m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c437321a778f4d81b6b83e0e143ff365 : SUCCESS in 1h 27m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1aee05bb98474efdacec50ea89a9a9ee : SUCCESS in 1h 13m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b734b6fc44eb4dbe94e0641d88a47897 : FAILURE in 21m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8cc074261c94406eacb4c307bd1d800d : SUCCESS in 2h 18m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/14e854dea79243ea8328a8d31c533495 : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c4cee47a20ec4f79a917c9c1770914ab : FAILURE in 40m 39s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84bed0bd6c0e4ad4ba951fa15c00542d : SUCCESS in 1h 02m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0baf88fb6c34849b09aba56af29267d : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d6b554886f3cd7fc2cd4c9575814dbfc5410d140","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-04 11:39:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"96ce9eadd799e8383fe10a88b7184f0325532384","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 14:28:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5f0e5f910923482690ce8a271b15b6e1 : SUCCESS in 6m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b1234924bad74d74b2f71ea1b5497f01 : SUCCESS in 1h 32m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/154034ae785440db916ababcd091a04f : SUCCESS in 2h 08m 42s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c53f4d6d77894962a0c7107c60f295b2 : FAILURE in 26m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/946189e1226242daa7d170354511d192 : SUCCESS in 2h 18m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9db3d881719341d5afb391180cd871a7 : FAILURE in 32m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6cf46178028463fb06df2904f12a95e : SUCCESS in 2h 46m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e3685a30189549949b438450797184fa : SUCCESS in 1h 19m 46s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/94bbcc2069b04749b51f11245dc0ac0d : FAILURE in 35m 46s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9356d0c9f4bd476aab3344989d4529e1 : SUCCESS in 1h 05m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8845407bd73c4980aa3c295f8f9b9320 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"7f6e272bd4e7f9286eabcba894dab547e4a5880f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-04 21:04:25.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c29cf8cf09c7b3d10d9e4c199fc9bc72c62ab4eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 23:36:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/26aa5127a15843bebdd7d22e9c4cd070 : SUCCESS in 12m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5e998fd29fdd44528912fad8abb6c3c6 : FAILURE in 2h 03m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8e5a49bb86c24504a277cc11f53f6851 : SUCCESS in 2h 28m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/148acdebd0b64ec48e06257e33b80f5a : SUCCESS in 2h 21m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e94b3ac442b8455bb0b577e5f0c0829b : SUCCESS in 1h 32m 37s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/15767ac6b23c4422a09b3c9ad1566216 : FAILURE in 25m 55s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca6176dc35904f4389819bd512403ec2 : SUCCESS in 2h 24m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/11eab55445eb46d38fb97aec900971ad : SUCCESS in 1h 21m 34s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b195b275ff314fe1a433f51472f53b24 : FAILURE in 44m 17s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/308cfe0433674976bf5e110d4fc7f88d : SUCCESS in 1h 25m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/724dc0b3ed69401b8df6da0b0d91fefc : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":2},{"id":"db11a346795dc30ca0a3c4f9734537c7de49a87e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-06 09:09:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2c934b19cc1dc40e1d7326cdb6f2ed8f0400a816","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 12:17:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4b1a96cace0e4db7ab4c35a32b616ff9 : SUCCESS in 8m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d0f5cf1ae2a344289c8afa78ffbfcebb : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b1d4f0ed779447f9bd92b83dff3a61be : SUCCESS in 1h 54m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/6857592f80514cd18142876f486db338 : SUCCESS in 1h 46m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1990abba74824301a06ab7b32f5afec5 : SUCCESS in 1h 44m 11s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ecd4c76b748b438eb6cf227d4e75f0f8 : TIMED_OUT in 3h 04m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8442e97087948e682d8c723396990e5 : SUCCESS in 2h 19m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9b7470f23aaa4bb79722215ea2ddb0a9 : SUCCESS in 1h 35m 38s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9fc87a07e95d46059311a259b11ec820 : FAILURE in 30m 50s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c470d9cb6094d85b7e5238b50f9014f : SUCCESS in 1h 07m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2e5fec6c6aa4011bf4b2db1870f8f87 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":2},{"id":"85d0a14c79da3cb131f8efddd1cc019cc493cd95","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-06 12:28:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"02bcea9ae39965690617c5c97eb16694ed38af85","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-05-06 16:42:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ccf027478b4aaa7cc6cb6ff084cdb51c30607117","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 16:43:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a81c3004c366cd4b9237671c9d9a152a778f55f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 18:38:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/aa181f5ab3dc41f49c549aeb34e5aa3a : SUCCESS in 7m 29s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/65d7e22b6c3d4520b2fbeafe567e7315 : SUCCESS in 1h 29m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8d85de4d0bad4b16bad8f6a2e8573566 : SUCCESS in 1h 50m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3fde881fc89f44678ac555edb7fb721a : SUCCESS in 1h 35m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b96192d0b23749848584c5b2f16bee00 : SUCCESS in 1h 30m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5c33e937e5c647c7acb590f466c4b162 : SUCCESS in 1h 26m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53934c99fc1843ffa1938082b9b7dc63 : SUCCESS in 1h 09m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2fc7ef57e54757b9929656d7d48748 : SUCCESS in 7m 35s","accounts_in_message":[],"_revision_number":2},{"id":"bb85793a81468c8e7b81607f4498b9e80b9d71cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 18:38:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9640ea5750c7f30bb42141bf9f3a505659b72c8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 18:42:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d2c06fa3c3c4a5b8b19320b1df6e8a4 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"482c01cd67967ded6b5ef3676295274d25dd3b68","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-06 18:46:08.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit a21bbf7be191edc53a0caf32ac2683c2ed06f34e","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"26637323d5411893af28a190595054e74baa4aec","revisions":{"95d19ea63b1656fc020ab835b9d0cde789e40b18":{"kind":"REWORK","_number":1,"created":"2020-05-04 03:07:05.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/54/725154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/54/725154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/54/725154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/54/725154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/54/725154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/54/725154/1"}}},"commit":{"parents":[{"commit":"7f53a9d0ad854ba2e63bd32d62c5b024edcca647","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/7f53a9d0ad854ba2e63bd32d62c5b024edcca647"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-04 03:06:56.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-04 03:06:56.000000000","tz":180},"subject":"Fix rados linking for py3","message":"Fix rados linking for py3\n\nThis patch detects correct lib path of gnocchi and use it for symlink\ndestination\n\nChange-Id: I7b88a053ac5d779bc53a32c1294e76e2233cdd9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/95d19ea63b1656fc020ab835b9d0cde789e40b18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/95d19ea63b1656fc020ab835b9d0cde789e40b18"}]},"branch":"refs/heads/master"},"26637323d5411893af28a190595054e74baa4aec":{"kind":"REWORK","_number":2,"created":"2020-05-04 03:46:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/54/725154/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/54/725154/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/54/725154/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/54/725154/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/54/725154/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/54/725154/2"}}},"commit":{"parents":[{"commit":"7f53a9d0ad854ba2e63bd32d62c5b024edcca647","subject":"Merge \"Cleanup py27 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/7f53a9d0ad854ba2e63bd32d62c5b024edcca647"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-04 03:06:56.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-04 03:46:43.000000000","tz":180},"subject":"Fix rados linking for py3","message":"Fix rados linking for py3\n\nThis patch detects correct lib path of gnocchi and use it for symlink\ndestination\n\nChange-Id: I7b88a053ac5d779bc53a32c1294e76e2233cdd9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/26637323d5411893af28a190595054e74baa4aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/26637323d5411893af28a190595054e74baa4aec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
