)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~727371","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I2ea86b5fdbb14fa9b6472df56407592d945d2bb1","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"I2ea86b5fdbb14fa9b6472df56407592d945d2bb1","subject":"[ussuri][goal] Update contributor documentation","status":"MERGED","created":"2020-05-12 18:52:45.000000000","updated":"2020-05-16 08:00:40.000000000","submitted":"2020-05-16 07:59:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727371-1589615965558-be6e565f","meta_rev_id":"cb89cf54d6dd72c709337cc15fa14128c732c40d","_number":727371,"virtual_id_number":727371,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-16 07:59:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-05-15 22:39:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-16 07:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-16 07:48:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 22:39:48.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-05-16 07:48:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-16 07:59:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e7cc04ca4657a7abaa9451eb5b1c26b3f63b363","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-12 18:52:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4d9852541c99f019effd8a897dc7404a8f8f857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 19:47:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0cd310e3059f40ca9be67e81d4b2fa07 : SUCCESS in 9m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aeac559fcfa4c479d674bdd016797dc : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":1},{"id":"07780aed367e610c4d844f617c8f0dcb5c838eec","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-05-15 22:39:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2459362470c5e9ee9d562fb485437a05fbb52f84","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-16 07:48:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"72fa59511fadee071703ae198670130dfff0cc1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 07:48:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2835763d02362f4ffabc589b1ed4c41171b3387e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 07:59:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8fc978864cb72e8049c90b65c6ff05a48a5fe5bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 07:59:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/27186826bc504f109d0270039713f510 : SUCCESS in 9m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aec302b71fb6409f807d46afd084f8b5 : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":1},{"id":"7bb1d5b1510ade9b66d0523bf6d1c9538578fa0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 08:00:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d28fb504a60d48a9be3cec2ebc58f945 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f3961ad1af5675f85cfd0bdbe00703310717707","revisions":{"5f3961ad1af5675f85cfd0bdbe00703310717707":{"kind":"REWORK","_number":1,"created":"2020-05-12 18:52:45.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/71/727371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/71/727371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/727371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/727371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/727371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/71/727371/1"}}},"commit":{"parents":[{"commit":"ea5802efb03cfdc51df30238051e98bd5d20aa73","subject":"Merge \"Fix rados linking for py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/ea5802efb03cfdc51df30238051e98bd5d20aa73"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-12 18:52:35.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-12 18:52:35.000000000","tz":180},"subject":"[ussuri][goal] Update contributor documentation","message":"[ussuri][goal] Update contributor documentation\n\nThis patch updates/adds the contributor documentation to follow\nthe guidelines of the Ussuri cycle community goal[1].\n\n[1] https://governance.openstack.org/tc/goals/selected/ussuri/project-ptl-and-contrib-docs.html\nStory: #2007236\nTask: #38554\n\nChange-Id: I2ea86b5fdbb14fa9b6472df56407592d945d2bb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/5f3961ad1af5675f85cfd0bdbe00703310717707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/5f3961ad1af5675f85cfd0bdbe00703310717707"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
