)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~727375","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Id71feb3c492343708cb21d3f25987af705cb388c","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa/cleanup","hashtags":[],"change_id":"Id71feb3c492343708cb21d3f25987af705cb388c","subject":"Cleanup after repo_build and pip_install retirement","status":"MERGED","created":"2020-05-12 18:55:13.000000000","updated":"2020-05-13 18:29:15.000000000","submitted":"2020-05-13 18:26:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727375-1589394405130-3f177800","meta_rev_id":"3c877a3d4b78dc580bf30c29318507a0c5b93816","_number":727375,"virtual_id_number":727375,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 18:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-05-13 10:37:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-13 07:24:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-05-13 10:37:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 07:24:23.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-13 10:37:39.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-05-13 18:26:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abf843f0b178c7b202357f92b3b15ad9b893b84a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-12 18:55:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4a2d51a141dc8e80ae3355a50d7bf65687b97d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 23:04:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e7f575a942ba4329af6eb6cb62e9e0ee : SUCCESS in 10m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/fe96a85feccc470ab70a2b915f246cee : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/95895327b8834ca88252b39c723719a4 : SUCCESS in 1h 49m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ca630098606047b6a99b7151eba4183d : SUCCESS in 1h 47m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbf27bca46fb49efb2ed783e45b64a82 : SUCCESS in 1h 36m 09s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e84f4f90ea80429b9dc00c6b2a87e957 : TIMED_OUT in 3h 04m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79a3f9b202134e1d9d7d8708e4361119 : SUCCESS in 2h 16m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2bd190cb96cd4a08a05ce94106a166ac : SUCCESS in 1h 36m 54s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e7d90c24e46449b4aaba3f9d11b4824b : FAILURE in 38m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66b2bd6b9a40444f9273d98ad8b547fa : SUCCESS in 1h 08m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fda69c4b4234b7dacd3fa19b2d29067 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"31d712883c6ea6a71082eaa8db8c986e2dfa2d9f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-13 07:24:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"23ed7f58fac63909aa1d4846c47b1efe677c84a2","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-05-13 10:37:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acaa4c3cd22c477a840b78b4198359c2b9875f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 10:37:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"efd3b3774ef01ce96771882ea9fa21ef70868019","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 12:53:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/68bcfa9d030a40ce8d9549959ce12b1c : SUCCESS in 8m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/17d06821a25d4324afdbc2aae84614c9 : SUCCESS in 1h 21m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/de7fffd6721747db8b997aaf0723f410 : SUCCESS in 2h 08m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/058186479ed74ab889a2e6ec75c4eaf5 : SUCCESS in 1h 41m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c291cbf6f81842d0b632c9e5eaaf3427 : FAILURE in 35m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d00b0e9acb3a4a2f99021cf84541f640 : SUCCESS in 1h 22m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81aa5fabfeee4188881fc103554c415a : FAILURE in 44m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4cd3bbbce75418b813b88bfd89195cd : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d900c24f2e024b87a01d8400b4b196dce3fb2715","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-13 13:16:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2f99ccb65afcd387bdab1055349de9104f2266fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:26:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f20d97b450f447c6aaf3e63ac0934af0 : SUCCESS in 8m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/31e8aa9ea3e84725bb16e280aa20f682 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/af1912692389401296995a76357167f8 : SUCCESS in 1h 52m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4182b933dd3c45c0bf8c426d55e5cf08 : SUCCESS in 1h 28m 29s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f3dc5328f20644d0924bf4d9f913355f : TIMED_OUT in 3h 04m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1194ff5262042498097b0359606c31a : SUCCESS in 2h 32m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e8063e24e10942dab1c163eb3046e6d8 : SUCCESS in 1h 14m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbaff06702e14d44afbaece911ddafa6 : SUCCESS in 1h 20m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c71bb0abe824f94bc40efa977592bd7 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6b8330091660764ecb6113171af2a983488ddaeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:26:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc753aeb6a002076c58714507ce624ee9187e78e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:26:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/76c8428eaef84bf5b1faf6cbaa8b32d0 : SUCCESS in 7m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/34d999d3fea54f09b9eae194c68f4360 : SUCCESS in 1h 22m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e458c14dabbc4edf983c84b29b9280f2 : SUCCESS in 1h 56m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c06338cc2ed4fe4bc84973d4226e337 : SUCCESS in 1h 26m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/182addd55d454859a8a9e9c3216183f7 : SUCCESS in 1h 23m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/744b8216471a4127827b8d460179a7e1 : SUCCESS in 1h 05m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/707b98f61c6040a29b31f60730833a0a : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e72a2ab303cce8355b09db36d6c779a6b8085a47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:26:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"da26226b2ea918b14272a572fc28fa93e23b5fc4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:29:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df422cf0b4314e85be6498659b51df70 : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d114af09d0d6df6385b0c03d2a9dda91c3682312","revisions":{"d114af09d0d6df6385b0c03d2a9dda91c3682312":{"kind":"REWORK","_number":1,"created":"2020-05-12 18:55:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/75/727375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/75/727375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/75/727375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/75/727375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/75/727375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/75/727375/1"}}},"commit":{"parents":[{"commit":"ea5802efb03cfdc51df30238051e98bd5d20aa73","subject":"Merge \"Fix rados linking for py3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/ea5802efb03cfdc51df30238051e98bd5d20aa73"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-12 18:55:04.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-05-12 18:55:04.000000000","tz":180},"subject":"Cleanup after repo_build and pip_install retirement","message":"Cleanup after repo_build and pip_install retirement\n\nChange-Id: Id71feb3c492343708cb21d3f25987af705cb388c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d114af09d0d6df6385b0c03d2a9dda91c3682312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d114af09d0d6df6385b0c03d2a9dda91c3682312"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
