)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~734277","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~stable%2Fussuri~Ib6ccf4102ffc2fc659554e31e480b22b1bfa72d9","project":"openstack/openstack-ansible-os_gnocchi","branch":"stable/ussuri","topic":"osa-collections-stable/ussuri","hashtags":[],"change_id":"Ib6ccf4102ffc2fc659554e31e480b22b1bfa72d9","subject":"Use ansible openstack collection","status":"MERGED","created":"2020-06-08 20:56:06.000000000","updated":"2020-06-24 14:50:41.000000000","submitted":"2020-06-24 14:49:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734277-1593010153123-f0b752bd","meta_rev_id":"61d051d5b6d659c7a628cbd37698a277716b87a3","_number":734277,"virtual_id_number":734277,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 14:49:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-06-17 19:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-17 18:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2020-06-17 19:01:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-24 12:44:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 19:01:48.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-24 14:49:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e8b763238c9cee054b7a49dbacbf1e09dfd293b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 20:56:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"176b03ea762806282da734fbf616e99b3c186541","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 20:57:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a6ab6d7d6a70582f440cad5cf06c727fd98a3130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 00:01:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/36aab2b70ef048e7bb55b44706b4e6e9 : FAILURE in 8m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8b7b77334c944fcd81837932fcb91ecf : FAILURE in 26m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f24a58cc124f440d872696024c02f09f : FAILURE in 33m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbd0ce68d2f4445dbf69b07e95f5d53d : FAILURE in 31m 11s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/976b356e137e40c39d53ae2b5c6b69d1 : SUCCESS in 2h 58m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cc7c5c4b1944f50a643d6d6afd55093 : SUCCESS in 2h 34m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ada8c520e4e346cab2d6f33a22f1e556 : FAILURE in 33m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e5527baa16a40a393f55b1c13bea34d : FAILURE in 32m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c4396cc1464de78a620b3729dec0c5 : SUCCESS in 9m 44s","accounts_in_message":[],"_revision_number":1},{"id":"182cb9e3963d1a212cc692f541cc9c1c01e99092","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-09 06:07:51.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4a9601b2ce3852bde65e50dbb6f1e0ed1fda4c5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 08:39:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c4c1ba2afbc14e77bebafed2f83f38fe : FAILURE in 12m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6295aa2656804419a19ab03a8bea8955 : FAILURE in 29m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bdf354a83d6b4776b60b658068ab4628 : FAILURE in 26m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54082297a9bc49bba44265c48d4329e8 : FAILURE in 29m 34s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d77a25d174594bb6bea2b15b9b385c8e : FAILURE in 1h 44m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5172bc7623dd4bf1abd398b436999e89 : FAILURE in 1h 16m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0873733cab4842c19246b48605d7c7c0 : FAILURE in 33m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c28a71f0c980442d870498e69cd16b1c : FAILURE in 23m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a32c91999a1a46959421f21099a450ae : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":2},{"id":"c1544be19a929833ec3176c508c5e2094846564f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-15 10:45:12.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a2f6d4d088e83ce3b302f67987eecbce24ec317e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 13:07:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/73f2b2ac29144119b277d575a32dc79a : SUCCESS in 6m 40s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8caef0ae94be4af39a1d5b773e7327e6 : FAILURE in 21m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6a69f71c1aa443ef93963d63f9d78cbd : FAILURE in 31m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2883e039a9354168b0282754b6610d42 : FAILURE in 24m 04s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/148da7289d6b4dc394e942d19ecc7082 : FAILURE in 1h 55m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69822a70b3f54517b314eb907f17e042 : FAILURE in 1h 16m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5378d0dd9d1f49a6b9524c210dcc5236 : FAILURE in 23m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/426a834ee5324b4d9f022cb70ce55c5f : FAILURE in 21m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5ba7de80bfb41de8b90b4f00fe8f10f : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"e0fb4cb8707847cf636eafd9b805cd40467db484","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 13:57:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"349c75e39d1f5998bfacf28057ea1ff11e2e3321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 17:11:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3b9b0922a6b2499995f2bfac3d505319 : SUCCESS in 9m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/24d09e2e22834920aa4a02fbeb90b2a1 : SUCCESS in 1h 17m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/20c92fdec9d94300aa4c0b4523930a4b : SUCCESS in 2h 05m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bcd2ffcf58948359f3937857d80f532 : SUCCESS in 1h 34m 24s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3b561e77b9f0489c8592cb665f6f2dc7 : TIMED_OUT in 3h 05m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef37a5f3524845d093226da27f1ef1ff : SUCCESS in 2h 37m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/de5b52d71ec44278888ae8a3a5fe2a00 : SUCCESS in 1h 35m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/716fb309a9ed47db8d50fc1601d61aa0 : SUCCESS in 1h 07m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c231de7ae344ebac8c1ddb3b217892 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":3},{"id":"dedde48e64576dbbc1d75b65ef04f74953fc6ddb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 18:20:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"eebbe780e05eafcf37f5517b59080f180482908b","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-17 19:01:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bbb11d17e6fcc13bc1344fa4d1d023ba5d227ef5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-24 12:44:12.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5652784e9bf884a8b4135f8600e52dec17117877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 12:44:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1a87b55fc0bada0de9f3a5ee596ca3fb5dc883bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 14:49:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9dc5090760844946ab7999cf0ee2f5dc : SUCCESS in 7m 23s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/73fbd9bd8c494c8ab21b35db6426bec4 : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ddfd20a2ea564197b231c37dfbd9f4bb : SUCCESS in 1h 56m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccba04c7037740e292330c4adb37e2f3 : SUCCESS in 1h 08m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2bfdfc666f1942ea82f78cec3fa8dd97 : SUCCESS in 1h 29m 16s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a7fcb0805f94218934e644592875fb7 : SUCCESS in 1h 30m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a155d156e22f42dea57c4e0b43f9922f : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":3},{"id":"31055a4b1ecedcb61b89e6baf5fb1377cc616e69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 14:49:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7d0395726ae7358a11593636c6a3d536d13c3f9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 14:50:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07723fbc8b247248d85f5dc5ef74b70 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b6aaf81023b0ae327bf4b2abfd33e2edfcf21711","revisions":{"60e5ddb4f6022450761049b918779e1350d2af2c":{"kind":"REWORK","_number":1,"created":"2020-06-08 20:56:06.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/77/734277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/77/734277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/1"}}},"commit":{"parents":[{"commit":"3016f32f57d708b096b161dc8afb6b013cadd4fe","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/3016f32f57d708b096b161dc8afb6b013cadd4fe"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:22:04.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-08 20:56:06.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ib6ccf4102ffc2fc659554e31e480b22b1bfa72d9\n(cherry picked from commit 4067c265420bc27062e21616323a9c269ea7a86b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/60e5ddb4f6022450761049b918779e1350d2af2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/60e5ddb4f6022450761049b918779e1350d2af2c"}]},"branch":"refs/heads/stable/ussuri"},"4b3199e3c1d19c06845b79bda2a19f9522ae8d1f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-09 06:07:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/77/734277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/77/734277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/2"}}},"commit":{"parents":[{"commit":"3016f32f57d708b096b161dc8afb6b013cadd4fe","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/3016f32f57d708b096b161dc8afb6b013cadd4fe"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:22:04.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-09 06:07:49.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/734419\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ib6ccf4102ffc2fc659554e31e480b22b1bfa72d9\n(cherry picked from commit 4067c265420bc27062e21616323a9c269ea7a86b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4b3199e3c1d19c06845b79bda2a19f9522ae8d1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4b3199e3c1d19c06845b79bda2a19f9522ae8d1f"}]},"branch":"refs/heads/stable/ussuri"},"b6aaf81023b0ae327bf4b2abfd33e2edfcf21711":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-15 10:45:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/77/734277/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/77/734277/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/77/734277/3"}}},"commit":{"parents":[{"commit":"3016f32f57d708b096b161dc8afb6b013cadd4fe","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/3016f32f57d708b096b161dc8afb6b013cadd4fe"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:22:04.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-06-15 10:45:09.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\nDepends-On: https://review.opendev.org/734419\nDepends-On: https://review.opendev.org/734271\nChange-Id: Ib6ccf4102ffc2fc659554e31e480b22b1bfa72d9\n(cherry picked from commit 4067c265420bc27062e21616323a9c269ea7a86b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b6aaf81023b0ae327bf4b2abfd33e2edfcf21711"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b6aaf81023b0ae327bf4b2abfd33e2edfcf21711"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
