)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~739346","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I56e873114cf85b29233d50a866ee925615dba70b","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","hashtags":[],"change_id":"I56e873114cf85b29233d50a866ee925615dba70b","subject":"Fix run task conditionals","status":"MERGED","created":"2020-07-05 09:07:50.000000000","updated":"2020-08-25 09:54:08.000000000","submitted":"2020-08-25 02:53:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739346-1598324015089-f63fd487","meta_rev_id":"942b9f0099753129838c7d7dfdb97a1323911cff","_number":739346,"virtual_id_number":739346,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-25 02:53:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-08-17 17:09:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-08 10:30:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-08-17 17:09:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-08 10:30:29.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-17 17:09:49.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-08-25 02:53:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83a8619d1ad8e7ce5d50e02e56baa66716d042d9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-05 09:07:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56d138560d56f51d4744b6d3a493acb6b6a72e93","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-05 09:18:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9918c3503b222a577e9e83ec8cbbafdc2e2630ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 11:39:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/27eb1403499d4a1da8e3aa2413fc79df : SUCCESS in 5m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/84c001b27e154a95bc1aa473ab885bc4 : SUCCESS in 1h 18m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4ea8a7abae7a4e768ecd0e7b2fe49f26 : SUCCESS in 1h 42m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e066c0f4e09f44438e1ccdbbf74d2217 : SUCCESS in 1h 18m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c691d1fc8224e75be0cd6be524da399 : FAILURE in 54m 54s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4c9ca4b54bf741a9bec7550d45a836c1 : SUCCESS in 1h 52m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef5fa9841e25461d94e895a7939b6334 : SUCCESS in 2h 19m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5d0e1a70f351483f8ddbcb5bfea6c691 : SUCCESS in 52m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2973b37d723f4b4bb614741860a4972f : SUCCESS in 1h 07m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c958f9f6e116414bbddbfc0c22333f5a : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"94f1458c5463938300216e32c24fcc981619c96c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-05 20:19:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"93a92da5a731808a58a74e9ba9b08c8913910208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 23:29:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d9bab3a2f420456fb42efeae6ccde532 : SUCCESS in 7m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f079af0b22fa4ec984ebd06eb82398d1 : SUCCESS in 1h 17m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d805b407798b49d58395f25ea7c99386 : SUCCESS in 1h 50m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e0b52266bef46b5b7abd79551a0cfef : SUCCESS in 1h 41m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12e69442a3c14e61bcda032cdae1dc08 : FAILURE in 1h 18m 57s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/05142d6090e2473483b35506e5175d7e : TIMED_OUT in 3h 06m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4888939ed654083b22eeecced7acb74 : TIMED_OUT in 3h 04m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/05921cb62af845f998979425720b9166 : FAILURE in 2h 03m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ff89b436a15486b9c7a088e3131013d : FAILURE in 1h 44m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0453aee26545d3bc4c093d9c0c341c : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":2},{"id":"ccc5514cf0816027141cf1cdb989b02299aee5d5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-06 20:29:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b7b69ecb72b63652bb1469fb0c3ae9f8902fe0fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 23:44:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6e2bbfc762a54360a24c8b194552f015 : SUCCESS in 7m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/373ec360c8e447f2bb72e94bc6aa28ba : FAILURE in 2h 03m 19s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4792c1398580414b97e2e5231258120d : SUCCESS in 1h 50m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b5d2e52650444efbbc054f99cb0074f : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8f3f63db163407d932462cc05541ac6 : FAILURE in 56m 46s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a920cdfca9a1475084413ddbd2219d67 : TIMED_OUT in 3h 03m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4e27d899259436780904c57cb1b28dd : SUCCESS in 2h 36m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d8bd6a2bcc3040f282084d26c0484853 : SUCCESS in 1h 12m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/043b53fb471b473fa92220a19262d762 : SUCCESS in 1h 09m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37890e42489a468fb5b7bea87212752d : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"43d6e3a7bd691a352e5f51b4e14eebf3bb8b1fa9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-07 04:02:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7cbb063f73529f9e1202a467ee44c088d89fd4c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 07:11:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5960e3abd6bf44e8b516034579ce6834 : SUCCESS in 14m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5e46bd6157ce441eaa7131ff49cb0df5 : SUCCESS in 2h 01m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eb16f052d47147ed8f872a50a5d9129b : SUCCESS in 2h 25m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee9a7f756eff4d629e0923b421387372 : SUCCESS in 1h 46m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a1629349fe24e46b4f82cd4de9cdc9c : FAILURE in 1h 31m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/92ebf9c126864ea290721ab5e1f01f66 : TIMED_OUT in 3h 07m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34aff1745a414fe5ab530ded406b5809 : SUCCESS in 2h 43m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/26d59fe6cf3f4ec095e05ccb960506ce : SUCCESS in 1h 38m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/573ed9ffc8194b34bba87cd673b78a5d : SUCCESS in 1h 29m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bab39ba057584f9aa22ae7d014e6d73e : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a65b7b126d8826ae55adac37d0f8dc957f2cb86b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-29 13:45:20.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4cb8f97961f114d5802498b7718027299f6abcdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:12:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4c4314b8845044509a7729d8a4596351 : SUCCESS in 6m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4357cbbd372743fcbc5278d8365615fc : SUCCESS in 1h 21m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4f9f426407254ec69a61404778c2c91b : SUCCESS in 2h 07m 51s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5e5b6bcbb2b44aa5b263726ed6a9591d : SUCCESS in 1h 40m 02s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66609694cd704874a03114a847ce5b70 : SUCCESS in 1h 44m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21d32dbf54a4404f8c6eaf4762a91bc5 : SUCCESS in 1h 55m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/183570f7b01c4a06a45d5568eb4ace39 : TIMED_OUT in 3h 03m 47s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a3f2b2cbeee4804827294fbc341dfb7 : SUCCESS in 2h 15m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4e4644af97e9458fbab778254919927d : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/32083e8f11514c149025be10d9da0ee7 : SUCCESS in 2h 01m 30s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/79bc2c234eaa430594ae01c0fd09f5eb : POST_FAILURE in 42m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a3764e4139448f0a44dfa68d1ed0caf : TIMED_OUT in 3h 03m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53a15af5c46545dd871f4a40e00a39dc : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"9617eaee8c111b2c768d8a83f161b76f38cb36f5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-07 03:40:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f66d301df16399ef5ec6f5fe2d607c5575c0b0ef","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-07 03:40:35.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e0d9dc5bcc6960bdea0364b2e81cd928854d4509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 06:49:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ea590fca64de4a7aa3b5cf743c74eb95 : SUCCESS in 7m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/af6c8b3ab64942d3940742490e41e5df : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7c69a513f1404a1cac501997340c1127 : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b921167efe4f41b0bc4f8e9523098443 : SUCCESS in 1h 53m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20109c7e44914825815f22d36e332eb1 : SUCCESS in 1h 22m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d449e8783dd74c5eae78c7a06bb88fce : SUCCESS in 2h 14m 06s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5026d3ca87d0455ca59dfe28a3a77093 : TIMED_OUT in 3h 03m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22312018b13c45c79824a3ad13a1c6fe : SUCCESS in 2h 37m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e2cb2e152624c659b3488b66c20810b : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d91b98fb7bb74ea288e9aca54c94284c : SUCCESS in 1h 31m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98a10e8b14404a42a7f0bf5fa42c151b : SUCCESS in 1h 17m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fceab96f0c3f42e28e773b12b98d65b2 : SUCCESS in 1h 24m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35aeb46e37b84f6cb783bb1b0384498a : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":4},{"id":"168c27e0eba5fc9b462e986d15bb0c33d4a336e1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-08 10:30:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"74b3796ec3269962e8f45c634ad453b890c6b26f","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-08-17 17:09:49.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"599c1f2fe5e492ab8d63ecd5c0a9e1286cc5de34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 17:10:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"eb3d423401f41d96cdc229acca80870d35fe822a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 18:52:57.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/98dae48de5df4daebb951b325e42d887 : SUCCESS in 7m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cbfc2aa08876401abcc68717b4bf2be5 : FAILURE in 53m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6e717a4b7b9b4be8b35fde73be0d767b : FAILURE in 1h 37m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/104849b36dc54b74be6bf6fa04afd981 : FAILURE in 1h 16m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbaf6ea8cc3e464ebcbc8f7d8b0616a2 : FAILURE in 1h 04m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44f22e6a34ba47fbbc8787653f630a55 : FAILURE in 55m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3a4c5ae50c1641568a3a0578bb77afb7 : FAILURE in 59m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/25c1ea8552cc4ce1a22efa189b56e539 : FAILURE in 54m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf9fad474e4d4da7a55ac95459c9bbf9 : FAILURE in 54m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/088d6e4e75cb45c6aebeb6cc4cb016f2 : FAILURE in 45m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3f629d54394e76a71ea781ec02f127 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":4},{"id":"b979ac5e7f19479d055b0688695687444ace0853","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-18 10:09:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4708004756f2b2d0cdd337ae6acb406f79971a7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 12:40:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/588653d7cdfb4c1aa551e0237c6d6c2a : SUCCESS in 8m 19s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/332339ec962a4f66b1b4774b53eb6fca : FAILURE in 1h 09m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ae5363008bf4f699747c58ce6592c27 : FAILURE in 1h 23m 26s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f5845f69ae1b4beeb629c7f4603ba8d7 : FAILURE in 1h 07m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f11c69f820fb48eaaca660584bfe3dd3 : FAILURE in 46m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/035713a09c8e433caa8c22ce6166c73a : FAILURE in 54m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8a2458056ab144668e0ee66b945daa2a : SUCCESS in 1h 56m 17s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97e5fd69b8e24d759d2f17fce4173136 : SUCCESS in 2h 22m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/89f25be08aa347fb992d8be1e90ba13d : FAILURE in 53m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/165f1ff24aab4bcdb71683126d5d6dcd : FAILURE in 38m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d25231c580a461e9dae705764d200de : FAILURE in 44m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/724f142a0a6349f59ac3b27b6b6e07fc : FAILURE in 45m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3965960ccf564e42b3de01ad5d282c28 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":4},{"id":"b5e479874d35899ca6abbb46fea3c158bc711df7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-24 21:31:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bf967c705e088e3ea52b902b6e0182183e9412e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 00:59:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/faaa4c94441a488d9c58f8f95e9bbe6c : SUCCESS in 6m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/79a1e2b16c5c4c319ad367faf5d92a3c : SUCCESS in 54m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0b7f8ccfa0414b68994593234c466782 : SUCCESS in 2h 03m 13s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/465f0b5bb1824d6dbf5839e6226bd161 : SUCCESS in 1h 32m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f44de29193b48fabac372f940167733 : SUCCESS in 1h 17m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/754acaf117284f54b16a1e77cf55dbf9 : SUCCESS in 58m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/feb9e609f39040128aa81a1c341fa2ac : SUCCESS in 2h 56m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe020c7abe814eca88a94fdfc8a69e13 : SUCCESS in 2h 18m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b5f06ad1836a4a85ab89d54208c36cf3 : SUCCESS in 1h 24m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fb4bb2c1afea450687f2a51e8c6bcdff : SUCCESS in 1h 22m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1619e98fb5fa48baa8b5a6a96651df10 : SUCCESS in 1h 28m 20s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5c66fe87e874c659ac3f0b8acc2ff2a : SUCCESS in 1h 03m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e32d14b4a89441deb5f74e099f77fe86 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":4},{"id":"de1fc7317e90dddb1461406505ae7e37f59e7e7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 00:59:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"23f04f4b893149efa02d5e8a3d3e70c5c5d3d972","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 02:53:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bedb5209bfbb49348ab338e599884ecc : SUCCESS in 7m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/41a0f51d2dfd442586c9bf32383755db : SUCCESS in 1h 50m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ea3b290f21b64ad088a2d0f43edcc0f0 : SUCCESS in 1h 35m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/aa2b127e8e3e487ebaacf22f71590179 : SUCCESS in 1h 34m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4b04e704f844af8b4af774a2d210dfa : SUCCESS in 1h 22m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c289b1f336c44359afb02d8674e47498 : SUCCESS in 1h 04m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0b1001ac206e4fc691751b274a64b6ae : SUCCESS in 1h 16m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0596760990f54c22a3c3dc01d67b1e49 : SUCCESS in 1h 10m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2609bf6813e64286ab03ab2512797f2c : SUCCESS in 1h 08m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5b72e73c83048f2b28c4f22f3bca1e7 : SUCCESS in 1h 04m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9965b9f66cb741259da963b7377c0e8b : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":4},{"id":"a9f4cd46f2959c1cd946dd52080e74ae0813c60d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 02:53:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5117f0396685fecd402a7c32140249e833f0d239","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 02:54:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d67e5190b0403c9d3dbbbc8813d653 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4},{"id":"942b9f0099753129838c7d7dfdb97a1323911cff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-25 09:54:08.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 278b3d36f3487401ef5067b99571fa12f40a8b55","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4fb7413e5717904cc1fe6db678c6a852b2842bde","revisions":{"cdf81ffc11b92b562dacd180e44c4617670a6906":{"kind":"REWORK","_number":1,"created":"2020-07-05 09:07:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/739346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/46/739346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/1"}}},"commit":{"parents":[{"commit":"95d02e709ade6ebedc38e125214b4bab64473d02","subject":"Cleanup ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/95d02e709ade6ebedc38e125214b4bab64473d02"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-05 09:07:38.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-05 09:07:38.000000000","tz":180},"subject":"Fix run task conditionals","message":"Fix run task conditionals\n\nCurrent conditionals didn\u0027t work correctly with running tasks with --limit\n\nWe should intersect with ansible_play_hosts to get tasks run for specific\nhost only (ie running with --limit)\n\nChange-Id: I56e873114cf85b29233d50a866ee925615dba70b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cdf81ffc11b92b562dacd180e44c4617670a6906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cdf81ffc11b92b562dacd180e44c4617670a6906"}]},"branch":"refs/heads/master"},"a698c6c2610b2effb57812ddd74b646cedf8ca03":{"kind":"REWORK","_number":2,"created":"2020-07-05 09:18:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/739346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/46/739346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/2"}}},"commit":{"parents":[{"commit":"95d02e709ade6ebedc38e125214b4bab64473d02","subject":"Cleanup ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/95d02e709ade6ebedc38e125214b4bab64473d02"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-05 09:07:38.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-05 09:18:03.000000000","tz":180},"subject":"Fix run task conditionals","message":"Fix run task conditionals\n\nCurrent conditionals didn\u0027t work correctly with running tasks with --limit\n\nWe should intersect with ansible_play_hosts to get tasks run for specific\nhost only (ie running with --limit)\n\nChange-Id: I56e873114cf85b29233d50a866ee925615dba70b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/a698c6c2610b2effb57812ddd74b646cedf8ca03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/a698c6c2610b2effb57812ddd74b646cedf8ca03"}]},"branch":"refs/heads/master"},"9c7513e67f5271b42bc7ad0deb764627c8c4a952":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-29 13:45:20.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/739346/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/46/739346/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/3"}}},"commit":{"parents":[{"commit":"99b671d98471ae4c91a3fea9da227bf2d4fc7c23","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/99b671d98471ae4c91a3fea9da227bf2d4fc7c23"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-05 09:07:38.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-07-29 13:45:20.000000000","tz":0},"subject":"Fix run task conditionals","message":"Fix run task conditionals\n\nCurrent conditionals didn\u0027t work correctly with running tasks with --limit\n\nWe should intersect with ansible_play_hosts to get tasks run for specific\nhost only (ie running with --limit)\n\nChange-Id: I56e873114cf85b29233d50a866ee925615dba70b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/9c7513e67f5271b42bc7ad0deb764627c8c4a952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/9c7513e67f5271b42bc7ad0deb764627c8c4a952"}]},"branch":"refs/heads/master"},"4fb7413e5717904cc1fe6db678c6a852b2842bde":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-07 03:40:35.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/46/739346/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/46/739346/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/739346/4"}}},"commit":{"parents":[{"commit":"801a24e3e45cef339990e5baf2e72bb7de578b22","subject":"Revert CentOS 8 jobs to voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/801a24e3e45cef339990e5baf2e72bb7de578b22"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-05 09:07:38.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-07 03:40:35.000000000","tz":0},"subject":"Fix run task conditionals","message":"Fix run task conditionals\n\nCurrent conditionals didn\u0027t work correctly with running tasks with --limit\n\nWe should intersect with ansible_play_hosts to get tasks run for specific\nhost only (ie running with --limit)\n\nChange-Id: I56e873114cf85b29233d50a866ee925615dba70b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4fb7413e5717904cc1fe6db678c6a852b2842bde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4fb7413e5717904cc1fe6db678c6a852b2842bde"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
