)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~739389","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Ief6fd82fbd2f208798237ac228f9904e2e6c0132","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"Ief6fd82fbd2f208798237ac228f9904e2e6c0132","subject":"Add centos-8 support","status":"ABANDONED","created":"2020-07-06 03:58:46.000000000","updated":"2020-07-07 05:05:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2fa5bbc5ca2ab93fd2e6d30847e42d6e0b0d85df","_number":739389,"virtual_id_number":739389,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 07:06:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1c6674d0e46f489e66d74bbf4ab3b87dd362987","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-06 03:58:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf1bb2242cd1088121592758df8187e0f20f3329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 07:06:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bf5418e997a344c2b45edd6883c4e6da : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/100cb696ff0a489c8a1d95f009a861c1 : SUCCESS in 1h 33m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/058deb69b85741b0bc7f81e210623aa3 : SUCCESS in 1h 48m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbd4cba5a76b45e7867cc3c4d7bd7c98 : SUCCESS in 1h 24m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b040ddbe3c894f3da5e81fdfebcda58a : FAILURE in 1h 10m 35s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b5b3693b63aa49ad9e7e3dc2f7f37a64 : TIMED_OUT in 3h 05m 32s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02118d54bade4d0fbc38b58ac0ae94f7 : SUCCESS in 2h 19m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c809560272c447ddbd674a9cbe531175 : FAILURE in 1h 01m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ae69d2fb92942598cf49ffc5d5f6cff : SUCCESS in 1h 14m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d5d152ff1e04690b24b1c4af03038e5 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":1},{"id":"bc1e21a84ff710aabe0f85bacc22b4571255f42c","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-07 04:57:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8375b19bbf2b8c11512f1f1be8ef04cad51652f1","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-07 05:05:16.000000000","message":"Abandoned\n\nNeed to fix something so removing it and resubmit again","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8cba1c5e4e599e58e37dd7c57fb331fd9bfba7d","revisions":{"e15c031f0b60835098879b96fc6e07bb935603ba":{"kind":"REWORK","_number":1,"created":"2020-07-06 03:58:46.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/89/739389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/89/739389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/739389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/739389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/739389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/739389/1"}}},"commit":{"parents":[{"commit":"95d02e709ade6ebedc38e125214b4bab64473d02","subject":"Cleanup ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/95d02e709ade6ebedc38e125214b4bab64473d02"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-06 03:57:30.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-06 03:57:30.000000000","tz":-240},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nRenaming redhat-7.yml to redhat.yml for centos-8 support.\n\nChange-Id: Ief6fd82fbd2f208798237ac228f9904e2e6c0132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e15c031f0b60835098879b96fc6e07bb935603ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e15c031f0b60835098879b96fc6e07bb935603ba"}]},"branch":"refs/heads/master"},"e8cba1c5e4e599e58e37dd7c57fb331fd9bfba7d":{"kind":"REWORK","_number":2,"created":"2020-07-07 04:57:42.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/89/739389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/89/739389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/739389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/739389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/739389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/89/739389/2"}}},"commit":{"parents":[{"commit":"95d02e709ade6ebedc38e125214b4bab64473d02","subject":"Cleanup ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/95d02e709ade6ebedc38e125214b4bab64473d02"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-06 03:57:30.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-07 04:57:34.000000000","tz":-240},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nRenaming redhat-7.yml to redhat.yml for centos-8 support.\n\nChange-Id: Ief6fd82fbd2f208798237ac228f9904e2e6c0132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e8cba1c5e4e599e58e37dd7c57fb331fd9bfba7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e8cba1c5e4e599e58e37dd7c57fb331fd9bfba7d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
