)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~742324","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I95c2c7c158cfcad6746188b24a0e06ffd2564244","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"I95c2c7c158cfcad6746188b24a0e06ffd2564244","subject":"Add centos-8 support","status":"MERGED","created":"2020-07-22 01:28:03.000000000","updated":"2020-07-27 18:38:04.000000000","submitted":"2020-07-25 23:51:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742324-1595721117580-85cd6830","meta_rev_id":"a3ae02dc638f58d8c8618e28543e1c1cc927af12","_number":742324,"virtual_id_number":742324,"owner":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-25 23:51:57.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-07-24 11:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-24 18:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-24 18:29:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-24 11:59:23.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-07-24 18:29:10.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-25 23:51:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a04e25c9acf2adcc001be481a8bcd3d647b6686e","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-22 01:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"793c7a951a07e13843c2d3c30ae64e6c694d58f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:44:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b147771acab84956ab10d1f83569e96f : SUCCESS in 10m 37s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/07cc0a2596cb41adadbb34bdbd98dc64 : SUCCESS in 1h 18m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/abd6aa467c6d4427a67d4ccd4d26c6a9 : SUCCESS in 1h 43m 12s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d751ff1154b44d35866e89813bd00d13 : FAILURE in 1h 08m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1106e239869941f397228a76bed8d17c : SUCCESS in 1h 31m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/877314f3469e46da964d4fcccd937c41 : FAILURE in 1h 03m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/15a73b8d528f4ac19a370dec63c7f7d5 : SUCCESS in 1h 54m 11s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65fe6de5816a4818bdcd4260fef6e943 : SUCCESS in 2h 10m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/40cee1477df8474eac2cb6ae45880320 : FAILURE in 1h 08m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e9ef758ccb2341b48db23867a374dde4 : FAILURE in 1h 05m 32s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f8542cdfecc44499d2a409c9e785b6b : SUCCESS in 1h 09m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec4facc6cbf463ab0413dd03ae660f4 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7c7b28bde75f511242aca6181c5417e06264cb4b","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-22 04:21:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d453eb0428f8ec41866218755f48674c8c525258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 07:19:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/348d87eec223428cb286fb2da2b48cae : SUCCESS in 6m 16s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3e1714faa96949afbc53c37be8625650 : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fbb54184d5f34b5e8bb05610907f848e : SUCCESS in 1h 28m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/328b381089a24f52a9c640b14c9c8361 : FAILURE in 1h 02m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a40b22cf42ea48418741b59c4a63c1a3 : SUCCESS in 1h 15m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/749f502391434af3b1db1a8b3bb5556b : FAILURE in 38m 33s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/911f9d6c260a415aab1cb2a6ebf67d65 : SUCCESS in 2h 54m 14s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/716c7b8ac31343ae9128ce53df79e0a1 : SUCCESS in 2h 09m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/965b04eeb4584e7393748cc50d8d4518 : SUCCESS in 1h 14m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3c11ea07a2b44ebebee8c3b261b13301 : FAILURE in 52m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96788aee8a44440391c5eec1a081bc73 : SUCCESS in 1h 06m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e1d0383aa114bf481c19ce6db1f07e1 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":2},{"id":"96c162dd40028aff016f28cfe3c3c8f6e440e269","author":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"date":"2020-07-22 17:00:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"29864610575474849a091c1b17479200168feb14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 20:12:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/924919ea2e104dc5b177fd82c5fef63c : SUCCESS in 6m 21s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a76ef8e3d5ce4f89a8dadf5202fdcd00 : SUCCESS in 1h 24m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ffc91971ffc9471490485ca9748cdd9b : SUCCESS in 1h 51m 01s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7c76fdaa2e8c477e92fb5ce169b10dbb : FAILURE in 1h 11m 43s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3d2f05d51bc4331832eebdc3dc59470 : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24e161e1e29f49e1adc95f3131b9f6d5 : FAILURE in 1h 03m 20s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d61cadb6ff4c4bf4adde9c3e124521b8 : SUCCESS in 2h 46m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2fda673bce44af9a1c7166b9786454b : SUCCESS in 2h 19m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4da82c970e53416996f9d70d62bb03de : SUCCESS in 1h 28m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6c5338f3b5174ab9be76c5c292ed19d6 : FAILURE in 53m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75a7ac79954f4da9b3b9a0fed64c64f9 : SUCCESS in 1h 07m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a275c7b8aad4494bef3a34b4c6f3396 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"d228274d0bc8956c551bc15a9bf6e918d2480b03","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-24 11:59:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1f010f6a7df05f5da11b0fe17560b7c3b79b84c8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 14:29:39.000000000","message":"Patch Set 3:\n\nHere I think all the devel packages are for wheel builds and should not be installed on the target host.","accounts_in_message":[],"_revision_number":3},{"id":"cf372ec050d26e195c75498de0f5d946e47b7dd8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-24 16:17:27.000000000","message":"Patch Set 3:\n\ntbh here I\u0027d do that with a follow-up patch","accounts_in_message":[],"_revision_number":3},{"id":"9277d25e195d9242ac2ccd7dc18964c1dae72b4d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 18:29:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b204e681c56e39fb9c33717df1790d0bf5bd2226","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 18:29:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6809a29c02a80be8fb7026ef21337bc796f7fa0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 20:32:04.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/003e1191761e42b6926b0d4d482526a0 : SUCCESS in 9m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bbeeb416308147dfb7f85bcdecc84307 : SUCCESS in 1h 18m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9c4e9eadf66644379b907284ee45565a : SUCCESS in 1h 41m 17s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/bc1f5ec7de50491bb526c2fadd95ea61 : FAILURE in 1h 56m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0fb76617b2a84cf88fd74b0c8db79b09 : SUCCESS in 1h 30m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ace27c2e1155449fb9ae61c6bd0a7ff2 : FAILURE in 1h 24m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f99f41614b4844d0b83beded53e18ecb : SUCCESS in 1h 13m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/017f90c9d0dd455190e4c9911eb4161f : FAILURE in 52m 29s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81f73e99988642b6b1ce00ddec9c1e49 : SUCCESS in 1h 07m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5b76cca0d564cf7a438b8cf4d3a20bf : SUCCESS in 6m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/88c631d11d45417c95e9089279f1cb3d : FAILURE in 1h 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9f1a9dfea920c92cf51b4f92fb60c7444a9b3a16","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-24 21:15:04.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c731207d0de47db809b3e5fdaf32336b45a68d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:55:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d4c7a7b2919a4e928947a05f7f27c3db : SUCCESS in 8m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0ab90e990fe84933b25ad1df7e238647 : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4e493d86df714ebf971e962ed4335e2b : SUCCESS in 1h 31m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c108e989182b4aa1a07326794542b24c : FAILURE in 1h 08m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53b02e9af8c24cae822c381dc563a303 : SUCCESS in 1h 17m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cab44204675d4581b31fb8fedf0cb8fa : FAILURE in 52m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7e2ece4bc695475dba58298ebce8d5ff : SUCCESS in 2h 34m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc13244c4f2948138c9b09ad4a98e48e : SUCCESS in 2h 06m 17s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a4ebc05cdb8649a882fea3228f678328 : SUCCESS in 1h 10m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ba4c73d2a5b54142976f4444eed2b7e4 : FAILURE in 49m 55s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f061009eb2b4b36828070a9e7550d2b : SUCCESS in 49m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c22ae3bc54400796b8c82f93f818d4 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":3},{"id":"a3fe754ad5e399f799b06189e0958d6d01f0a5c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 23:56:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70868ebcbdd5655a39e01c52195b547a1202d3f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 01:50:17.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9e7aea7ff96e47eb9c1a6ee4284215b4 : SUCCESS in 7m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1cc6b3f728b346988c9d64f0b0d9da96 : SUCCESS in 1h 10m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a4d623ec4132422bb57ffce629d01e18 : SUCCESS in 1h 45m 22s\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/05f4ed1bb0fc498cbac8c9b54c9c074c : FAILURE in 1h 46m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d125b0de5b34ffaac73d5eaa7dbac2c : SUCCESS in 1h 26m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b86d0cbf1e1f4e12b65822549235a349 : FAILURE in 1h 02m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9e02ee9043a7463ba84c7eb5ef7785f6 : SUCCESS in 1h 15m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/adc9544ff5c64bee81c33dc784bd1e2d : FAILURE in 46m 16s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4673c72eebce4b2a8e2867b67c540d8f : SUCCESS in 1h 06m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27adaa5bb9b74f57849bdfdc1c761a50 : SUCCESS in 6m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/df24025cae074ca3a1bb98cccc52ca7d : FAILURE in 1h 11m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d82e7df66ceb3f1751df9462ddef2c86d111a04","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-25 19:23:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e34290e25d08d9c324eb8862b2add171a5b33dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 22:22:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c872b96b0c9c429f804681a8f5699f4e : SUCCESS in 7m 34s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cfc9c07f200d41ee944c0f46483f54c0 : SUCCESS in 1h 12m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4b9da992737a4ce78365e122d164af4b : SUCCESS in 2h 07m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fc7b092cd2c740669f70c0c6294ec863 : SUCCESS in 1h 32m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ed2d9a87a9f4b3b851790ac4bc6985d : SUCCESS in 1h 23m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4a9cbd7867a4ab7b708816358750dd5 : SUCCESS in 1h 15m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c8fc27b836d7401bae0403a3230bdea6 : SUCCESS in 2h 56m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e49b93176f744f5a5fd989884fd76e0 : SUCCESS in 2h 09m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7543e57184734cd3983cc6c52a007969 : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/50c31745020b4433b7e02c211c4b206f : SUCCESS in 1h 10m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eee11cd9f38844f3bd3c457f9e42b121 : SUCCESS in 1h 07m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/480373ba83ff40c09eb92dbcbe3d8e78 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":3},{"id":"f3e595cff05beb9a9f9a031f4acbf69bdd865c55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 22:23:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fb50b691028de6c7cd97669ce4133fd70f668919","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 23:51:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"1e3b5d8879ffe507ca7734122b4667a146f3bf9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 23:51:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3b8f635cd1ef48ed84a5533e6a537ae3 : SUCCESS in 7m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3cd072170aad43bfa70ed8069e8dd40c : SUCCESS in 53m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8d4575eab32f4adb9ef6743a81bee744 : SUCCESS in 1h 05m 23s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1487b22884fe4deeadf88bd33e122743 : SUCCESS in 1h 05m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6661583a338f48c997c4bfd1903d6ec7 : SUCCESS in 1h 21m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59397dc9281e4728abbdb90b0dbdee87 : SUCCESS in 54m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a6682ac0abe940fc837b999e47ca3f32 : SUCCESS in 51m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d4cf7d0f404140adaed4fa6794a2c366 : SUCCESS in 50m 42s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c1fe61c4fbe48a3a2c5c4964d8e36b8 : SUCCESS in 1h 12m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f2b86a784e413abb6c83618dec959e : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"51c8417752d26e046a97f362abf2e3334defd412","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-25 23:53:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd11ac187a1c46fbbe7dfb1e1dc34624 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":3},{"id":"937f629587cd0b43f355f5374745179b25ac45e8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-27 18:38:04.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 38fedc6c6896755e52ab0ba0b7cbd999dbb4d903","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"99b671d98471ae4c91a3fea9da227bf2d4fc7c23","revisions":{"2e0aa083c14deb4275dbcd4e3cd23a8cfeaac7c6":{"kind":"REWORK","_number":1,"created":"2020-07-22 01:28:03.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/24/742324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/24/742324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/1"}}},"commit":{"parents":[{"commit":"95d02e709ade6ebedc38e125214b4bab64473d02","subject":"Cleanup ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/95d02e709ade6ebedc38e125214b4bab64473d02"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-22 01:27:36.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-22 01:27:36.000000000","tz":-240},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nRenaming redhat-7.yml to redhat.yml to support centos-8.\n\nChange-Id: I95c2c7c158cfcad6746188b24a0e06ffd2564244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2e0aa083c14deb4275dbcd4e3cd23a8cfeaac7c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2e0aa083c14deb4275dbcd4e3cd23a8cfeaac7c6"}]},"branch":"refs/heads/master"},"e3f01b9f1992e6ef61bee07e0fc10ba6513c2ec7":{"kind":"REWORK","_number":2,"created":"2020-07-22 04:21:38.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/24/742324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/24/742324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/2"}}},"commit":{"parents":[{"commit":"95d02e709ade6ebedc38e125214b4bab64473d02","subject":"Cleanup ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/95d02e709ade6ebedc38e125214b4bab64473d02"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-22 01:27:36.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-22 04:21:12.000000000","tz":-240},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nrenaming redhat-7.yml to redhat.yml to support centos-8.\n\nChange-Id: I95c2c7c158cfcad6746188b24a0e06ffd2564244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e3f01b9f1992e6ef61bee07e0fc10ba6513c2ec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e3f01b9f1992e6ef61bee07e0fc10ba6513c2ec7"}]},"branch":"refs/heads/master"},"99b671d98471ae4c91a3fea9da227bf2d4fc7c23":{"kind":"REWORK","_number":3,"created":"2020-07-22 17:00:00.000000000","uploader":{"_account_id":28752,"name":"Satish Patel","email":"satish.txt@gmail.com","username":"spatel"},"ref":"refs/changes/24/742324/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/24/742324/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/24/742324/3"}}},"commit":{"parents":[{"commit":"95d02e709ade6ebedc38e125214b4bab64473d02","subject":"Cleanup ansible_python_interpreter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/95d02e709ade6ebedc38e125214b4bab64473d02"}]}],"author":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-22 01:27:36.000000000","tz":-240},"committer":{"name":"root","email":"satish.txt@gmail.com","date":"2020-07-22 16:58:43.000000000","tz":-240},"subject":"Add centos-8 support","message":"Add centos-8 support\n\nrenaming redhat-7.yml to redhat.yml to support centos-8. (setting non-voting to break telemetry dependency)\n\nChange-Id: I95c2c7c158cfcad6746188b24a0e06ffd2564244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/99b671d98471ae4c91a3fea9da227bf2d4fc7c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/99b671d98471ae4c91a3fea9da227bf2d4fc7c23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
