)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~743541","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I9d356bfbf18c370c365937114c18dad92fa4ea53","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa/el8","hashtags":[],"change_id":"I9d356bfbf18c370c365937114c18dad92fa4ea53","subject":"Revert CentOS 8 jobs to voting","status":"MERGED","created":"2020-07-28 13:47:55.000000000","updated":"2020-08-05 07:06:17.000000000","submitted":"2020-08-04 18:26:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743541-1596565605153-76a6aaaf","meta_rev_id":"e2c6f775689bebdc80ee3373414c1d472c2cc023","_number":743541,"virtual_id_number":743541,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"date":"2020-08-03 12:58:14.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-04 18:26:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-08-04 16:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-03 16:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-08-04 16:23:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 16:55:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-04 16:23:03.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-08-04 18:26:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"216620f77fc195f6918be6a662ed5e4c16b54e75","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-28 13:47:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee949925a564fde29c6bfbbbb7882fc266cf508c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 17:23:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/88252a3e51f24c0db66f2135da89b0f2 : SUCCESS in 9m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f8fa4ea04be7411eb9f5b2b18a222244 : SUCCESS in 1h 26m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/854ba8d0add947cd926ad12893b41c15 : SUCCESS in 1h 09m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/153f00a6f16049d491d5cd897934b8f3 : SUCCESS in 2h 01m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd53fd204e754b06ab71f0a70b8d14f1 : SUCCESS in 1h 32m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8679664e1c447c98a1a72ab569ac14a : SUCCESS in 1h 22m 47s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a4ba07d567ce490a81821193021d329b : TIMED_OUT in 3h 03m 29s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/881a123d88054cf98f1fcfd8ad7357c9 : SUCCESS in 1h 46m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dfe77553a8cc4c83a87e44c3891f2f62 : SUCCESS in 1h 23m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f27e4c855d514105aef364b6b8f08643 : SUCCESS in 1h 10m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b8ad18ace334a4eb5dfd56479dd4bd1 : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2809d0abb81c483f9c523f54b6c513a0 : FAILURE in 49m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308e9de235f448d7a8972871c334260a : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"049589db263576cb46956711da4ec81ff194f7ce","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-29 16:23:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9c0bba593664e20b3a691fc63915c1d252eab47b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 20:03:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a2e5cf43125e497aaf749320f0581461 : SUCCESS in 6m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5e3abecbcca1408481f2c27362b45937 : SUCCESS in 1h 47m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cf544824dc9e488bb4e876d02d2ab4b5 : SUCCESS in 1h 54m 56s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/531d06ffffcd4bf39868c29885d0d8d6 : SUCCESS in 1h 29m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c679959e8a29438288a84f68d4750731 : SUCCESS in 1h 57m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/212ec2ec006945ad95769cf6ba0303c7 : SUCCESS in 1h 18m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/61f1182351424bbdbef989b2821a0a1e : TIMED_OUT in 3h 04m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f26df4229ff1432b8db04e42676dafba : SUCCESS in 2h 04m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/250fe20bbf2247f4977ef0385f0ef222 : SUCCESS in 1h 10m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a93df62a1ce14e4a90027bb1781b0030 : SUCCESS in 1h 06m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/589dd34da69c4752923e51cb1a91ec6b : SUCCESS in 1h 07m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4d84f165b9d4d4c983f317bf82a9de4 : FAILURE in 47m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65c34e9d1cae49f49542b4a3db7e291e : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"61a0f49a077f20f1c18f955888365e5f82befdfe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-03 12:58:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ccd77240f1f685d26f30d8b56b700714478f06a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 16:13:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dacee6091d2441f9b3fbb7d16795a6c0 : SUCCESS in 6m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/20c2c3db1a874f73a2ac2c3c3f490179 : SUCCESS in 1h 25m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ad144413cc254c0eab22368ab47fef62 : SUCCESS in 1h 44m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b229dcf2057147f9837bce095c1104cc : SUCCESS in 1h 35m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe85723c7a2c43cfa33cb47d9cddd14a : SUCCESS in 57m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aa0703bc9e345a597d2751d8c0d476f : SUCCESS in 1h 14m 55s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/951495792e1a481ebb94e1a5881924cb : TIMED_OUT in 3h 03m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afb5e79e241a40b396d2325fbfcf4651 : SUCCESS in 2h 18m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0a1d5260eb3644dc894e7aa39343b968 : SUCCESS in 1h 19m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c6d582aeb6be4ace8083651383508e62 : SUCCESS in 1h 09m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e128a898f10b4ac5b6ee88009d5a0abe : SUCCESS in 1h 34m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1653fea4e904f7c8b4454bd547c6412 : SUCCESS in 1h 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce35a9733e843e399df916b687b3053 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"7aeaecfd9e476c85e2bd0fe2aefe0f16e119e5a0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-03 16:55:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"789820cc8628bd6d2e6b4277c13b74d9d7127dab","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-08-04 16:23:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c64069cd2b66687f69921d097ba846ede8b144d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 16:23:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"448fcbcfde06d5e73c9ab945ac402fe3210a9537","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:26:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b2e1a0182dd848b9bdb695c55a65cdc9 : SUCCESS in 7m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/64e8fc64a96b48f3b75af4789e2e3a8a : SUCCESS in 1h 19m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/df07454e3aab48f09f88ad0faa2c1fec : SUCCESS in 1h 50m 37s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0b07a81c48a44cea9ac5b9a94a7e7ed1 : SUCCESS in 1h 36m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c43054757592468daa9921253d69d9d4 : SUCCESS in 1h 28m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb6c2d865a4f494f9f6df5c7383069c3 : SUCCESS in 1h 38m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3518e62416644b5d8296824213725929 : SUCCESS in 1h 13m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b052683e24064893ba76632f770466de : SUCCESS in 1h 09m 09s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9474b299521e4ff09b1688c9155820e6 : SUCCESS in 1h 09m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac26ff921b7d4ddab640d78c0aa6600d : SUCCESS in 1h 20m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa98237505a4ab985bf427f66093c6b : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"a0dd0f875ecb8e78c46b041b4e944b75b585bfdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:26:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"42b684aa92ea049c1beb7bd80a60d7b9597f082c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 18:30:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ce45d7245c48cebb2b1dbecc950ae1 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1},{"id":"954239f53dee4271d798849e8c3b70b6f2b5402e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-05 07:06:17.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 6bcbe97e1d7ace79efe2bdb73345a7834cf4d68e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"801a24e3e45cef339990e5baf2e72bb7de578b22","revisions":{"801a24e3e45cef339990e5baf2e72bb7de578b22":{"kind":"REWORK","_number":1,"created":"2020-07-28 13:47:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/41/743541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/41/743541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/41/743541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/41/743541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/41/743541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/41/743541/1"}}},"commit":{"parents":[{"commit":"99b671d98471ae4c91a3fea9da227bf2d4fc7c23","subject":"Add centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/99b671d98471ae4c91a3fea9da227bf2d4fc7c23"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-07-28 13:47:48.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-07-28 13:47:48.000000000","tz":180},"subject":"Revert CentOS 8 jobs to voting","message":"Revert CentOS 8 jobs to voting\n\nChange-Id: I9d356bfbf18c370c365937114c18dad92fa4ea53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/801a24e3e45cef339990e5baf2e72bb7de578b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/801a24e3e45cef339990e5baf2e72bb7de578b22"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
