)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~747885","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~stable%2Ftrain~I56e873114cf85b29233d50a866ee925615dba70b","project":"openstack/openstack-ansible-os_gnocchi","branch":"stable/train","hashtags":[],"change_id":"I56e873114cf85b29233d50a866ee925615dba70b","subject":"Fix run task conditionals","status":"MERGED","created":"2020-08-25 09:54:23.000000000","updated":"2020-08-26 10:32:43.000000000","submitted":"2020-08-26 10:30:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747885-1598437817392-98d6fafc","meta_rev_id":"bddc01ac41adb0ed84bf432ffc41f634959cf960","_number":747885,"virtual_id_number":747885,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 10:30:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-08-26 08:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-25 20:02:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-08-26 08:15:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-25 20:02:25.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-26 08:15:26.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-08-26 10:30:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f683fb2a6a78962369a57e6c43c93f9f4bcbe7cd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-25 09:54:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"1483ffe741e053d647a818c2fa09000d7f6daac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 13:41:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bfbf1ac786d54eacb6ea44064ad3e50e : SUCCESS in 7m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cf584182e8bf4cea82fdc0e0eb37d615 : SUCCESS in 1h 15m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/71067847a57c41d5957474c8501aa238 : SUCCESS in 1h 05m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a70cb6c5652144b5a1c0db62acf51a1b : SUCCESS in 1h 52m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/60cdbda64adf4558a554b555d05d057d : RETRY_LIMIT in 14m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4fd988b8b614b3f8e483e88ffd6c3fd : SUCCESS in 1h 15m 10s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/03ba8d3578e144c19833f2bef8cbfb3e : SUCCESS in 3h 05m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23fe4e96723b42f58af258205103087e : SUCCESS in 2h 19m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0de7d8e9b5de43a9ac44ca3fec0e64b6 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/bcda35aa3fe24f4fb421773ce96f026a : RETRY_LIMIT in 21m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/26737209688b4e3f9588ec5a50781001 : SUCCESS in 1h 05m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/984feccefdec4864b16d304012abd3d1 : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":1},{"id":"320f76497e820e90eb39e3d8861a25b5a813e447","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-25 20:02:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3efac6612e3d87120432215b52eccf123a11eab9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-08-26 08:15:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a13ce4da59dde27cf5ba3512026386df2ad290a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:15:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfee0515ca58dc334b41c5c7374d34ad07db6604","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:30:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"17e045d846898f171b258762e257727cb186dedf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:30:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e09c6c7cd65b42b1875d55720bc94ea1 : SUCCESS in 6m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e1f9a2590dae4b3686ec0892089c90c0 : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5b3e629c506e4f3093741b8644a86035 : SUCCESS in 1h 24m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5e26c45e474243e4bded7fee65d0ff0f : SUCCESS in 1h 44m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/71764623a3b14e8fbe26f6f82e3a97e5 : RETRY_LIMIT in 22m 45s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c78383a8b5ed472eafd2ad715ec21fb3 : SUCCESS in 1h 22m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a232e205a864ac5a0eef4da5fc29064 : SUCCESS in 2h 02m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c640043001e04366bb22a1cf9c9b19c3 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72003811549b4a2fb412c9b5a58ef7d3 : SUCCESS in 1h 06m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7b38a33e7dd43dcb53e4aaa872ab4a8 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"bddc01ac41adb0ed84bf432ffc41f634959cf960","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 10:32:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3ad49f993a414aa89553c4885ce156 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fca7782578fa79bfddeeb6ea6c3e1478cd06c77","revisions":{"6fca7782578fa79bfddeeb6ea6c3e1478cd06c77":{"kind":"REWORK","_number":1,"created":"2020-08-25 09:54:23.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/747885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/85/747885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/747885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/747885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/747885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/747885/1"}}},"commit":{"parents":[{"commit":"762b85fbbfb47b1f182697a93a9ea3f2e11211d0","subject":"Fix rados linking for py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/762b85fbbfb47b1f182697a93a9ea3f2e11211d0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-07-05 09:07:38.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-25 09:54:23.000000000","tz":0},"subject":"Fix run task conditionals","message":"Fix run task conditionals\n\nCurrent conditionals didn\u0027t work correctly with running tasks with --limit\n\nWe should intersect with ansible_play_hosts to get tasks run for specific\nhost only (ie running with --limit)\n\nChange-Id: I56e873114cf85b29233d50a866ee925615dba70b\n(cherry picked from commit 4fb7413e5717904cc1fe6db678c6a852b2842bde)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6fca7782578fa79bfddeeb6ea6c3e1478cd06c77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6fca7782578fa79bfddeeb6ea6c3e1478cd06c77"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
