)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~751272","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I69c08fd522c1c80356000b1cf2a48a9b63327b3c","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"run_condition","hashtags":[],"change_id":"I69c08fd522c1c80356000b1cf2a48a9b63327b3c","subject":"Define condition for the first play host one time","status":"MERGED","created":"2020-09-11 13:14:24.000000000","updated":"2020-09-21 09:37:51.000000000","submitted":"2020-09-21 09:35:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751272-1600680946735-b2ad87a5","meta_rev_id":"07e9cfe5990309d7dc92700724212856011b6c11","_number":751272,"virtual_id_number":751272,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"date":"2020-09-16 21:29:15.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-21 09:35:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-09-21 07:15:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-18 06:39:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-09-21 07:15:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 06:39:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-09-21 07:15:52.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-09-21 09:35:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10c875a4fc4d6ce9c83983fc7075a8f769a078bb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-11 13:14:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfeb54df2b4073f8411d806f1a3ffecdae49606d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 16:15:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2a67ede6a82e498692f97b72271a8d0d : SUCCESS in 7m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2737a07430154144bdc448fd5356706c : SUCCESS in 1h 01m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b3b5dbf9e05445c18e6f64af289e7e6e : SUCCESS in 2h 01m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f9c0dfd40e1439ea8b0c4b5cca7ba91 : SUCCESS in 1h 28m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01058c47e2e4422383834b5b42b0c3b2 : SUCCESS in 1h 12m 06s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5a3ad8a7a3a44d1783b8f37606a8efa8 : SUCCESS in 2h 47m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23e1d06e2f124f19b919a7aee74ad077 : SUCCESS in 1h 55m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ac52c8ac074841f6bdaeb377cda8b017 : SUCCESS in 1h 12m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/531151d01d77466b9e92f30ad91c9f93 : SUCCESS in 1h 14m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b94a21ab4b948419f7c4bf8c4ef314b : SUCCESS in 16m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ba3bd1e6ebb15b9f9e01ca6f07e469baca22c29e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-11 19:18:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77aa9a85e9e5443ec5cb011e395133c01e12a539","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-15 15:59:17.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"69fb08ebf8352251bcfe7b4c67c435d71baac727","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 11:55:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3e2dc2cfd7021745420d1adfd865b142fde7b406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 14:51:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b93f340e361e44c083fafdfdd39d4cce : SUCCESS in 7m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4bf52c98b329490387747b05586feb0b : FAILURE in 1h 51m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f15e0df5fba84fceb97d189359ec3005 : SUCCESS in 2h 12m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c69a6cb7c06e4f8c979491c31f594ec6 : SUCCESS in 1h 29m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9620609bc5644e6a3f7a95b93274872 : SUCCESS in 1h 36m 10s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0c788ed916af4aaa840d0a98c05bb669 : SUCCESS in 2h 34m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2377439bfe904e0dbdf79dbee3a7cae9 : SUCCESS in 1h 58m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1df9094f87a24c70a4cf8adffbdf52f9 : SUCCESS in 1h 09m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a2dcd2ddfae472ba487ddd1ed7b7434 : FAILURE in 57m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/972aa2adecb340bca23dc974f59338b5 : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":2},{"id":"ec5fb3c9b46091e2d1d8d62253ab882ef3dc2f0b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 21:29:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"14772247bc7ed5ce656b045c5a9f6ffba4168e01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 00:40:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fb98f3b26cfc4f04922a592bcf9781ae : SUCCESS in 7m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/607a9e127cfc44b093150edc4e8b511c : SUCCESS in 1h 11m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fdccd28583eb48118b34ed57962baa33 : SUCCESS in 1h 29m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/858c989acd294ade89ffc842c14e21ce : SUCCESS in 1h 31m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bdeaae5b1c346d2a55bebcc7fbaa64b : SUCCESS in 1h 09m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0cffae4616de41a98eb3c18e7287ee3a : SUCCESS in 2h 51m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66cbbd554b6f44e3b582d28c599452d4 : SUCCESS in 1h 58m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2a16e21b33a4455f802696a5b5727914 : SUCCESS in 1h 20m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a9617b1a74f419aa905d353afb90585 : SUCCESS in 1h 07m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4caaf2e732c74749bab6602008d4eaea : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6f6980ffe7c0c2e69b57935df32855d8317e56f0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-18 06:39:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3d7e684df3238af7a41c6d84bee9013ccfabb5fb","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-09-21 07:15:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f8461b2adbc225abfebf10aca84caab2a2db3eac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 07:16:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7bf86243f3d550c368baf3beae71402eace5bbd8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:35:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2ac7c6a93e81bb779f2b670d7bb1aeaf35cee877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:35:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/15eaa123ae2648368f18119c9cee26b9 : SUCCESS in 8m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cd7fe8c274a44759a98403cad35ccdca : SUCCESS in 1h 19m 32s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b2ae8764da994a4da0619d3e01da5262 : SUCCESS in 1h 39m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60030a1e29bd469cbc2ef55a0c31592a : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d62aa1fdcb54c44b4ab6835fe1759a3 : SUCCESS in 2h 02m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1b7bb35a4bdf45ddb46b50183c86232c : SUCCESS in 2h 04m 05s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/750e30ef595f4bd2bc088170f23e4db8 : SUCCESS in 1h 05m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaabf5ee7ec648f298d17a39fe9a62c9 : SUCCESS in 13m 19s","accounts_in_message":[],"_revision_number":2},{"id":"d936da7f18cd52768515c7ccfbf947f71e45b39e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 09:37:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6cc28ad33694ff18a49af16a67347d6 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"acce40d693502181fef99d032e6f545d7d7ade1a","revisions":{"aa3b6315ec0cc6948c620da9b6ddd1eda90e4527":{"kind":"REWORK","_number":1,"created":"2020-09-11 13:14:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/751272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/72/751272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/72/751272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/72/751272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/72/751272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/72/751272/1"}}},"commit":{"parents":[{"commit":"c293c91eeb32390e383085f6e706a20fd5fc64e6","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/c293c91eeb32390e383085f6e706a20fd5fc64e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-11 12:55:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-11 13:14:04.000000000","tz":180},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: I69c08fd522c1c80356000b1cf2a48a9b63327b3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/aa3b6315ec0cc6948c620da9b6ddd1eda90e4527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/aa3b6315ec0cc6948c620da9b6ddd1eda90e4527"}]},"branch":"refs/heads/master"},"acce40d693502181fef99d032e6f545d7d7ade1a":{"kind":"REWORK","_number":2,"created":"2020-09-16 11:55:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/72/751272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/72/751272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/72/751272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/72/751272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/72/751272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/72/751272/2"}}},"commit":{"parents":[{"commit":"c293c91eeb32390e383085f6e706a20fd5fc64e6","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/c293c91eeb32390e383085f6e706a20fd5fc64e6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-09-11 12:55:40.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-16 11:55:55.000000000","tz":0},"subject":"Define condition for the first play host one time","message":"Define condition for the first play host one time\n\nWe use the same condition, which defines against what host some \"service\"\ntasks should run against, several times. It\u0027s hard to keep it the same\nacross the role and ansible spending additional resources to evaluate\nit each time, so it\u0027s simpler and better for the maintenance to set\na boolean variable which will say for all tasks, that we want to run\nonly against signle host, if they should run or not now.\n\nChange-Id: I69c08fd522c1c80356000b1cf2a48a9b63327b3c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/acce40d693502181fef99d032e6f545d7d7ade1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/acce40d693502181fef99d032e6f545d7d7ade1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
