)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~758746","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I8b8102da999ba06d1ab7039db029390726027380","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I8b8102da999ba06d1ab7039db029390726027380","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-10-19 09:19:58.000000000","updated":"2020-10-19 20:27:44.000000000","submitted":"2020-10-19 20:25:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758746-1603139145426-e0ec6abb","meta_rev_id":"2b6aa6a277c506651f38dcbd05977712c037f2c8","_number":758746,"virtual_id_number":758746,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 20:25:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-19 12:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-19 12:55:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 12:55:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-19 20:25:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbe6f13eb8d5e72527873731c676b12041d19aac","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-10-19 09:19:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f17b0df3b4fbc0b7462d115d2e5a91c8f9ebb70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 12:37:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/26587b94bf2b46cfa05ec16085a79403 : SUCCESS in 9m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3152c83fe2ed4dc6b7e7c2961138cd92 : SUCCESS in 1h 24m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/400e8c9547c64872a0c1452f34cb71ab : SUCCESS in 1h 56m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9863ea698e424c11a7bec67bfd7b95db : SUCCESS in 1h 37m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ef8a0c919714ec982d878d6f4e7616f : SUCCESS in 1h 24m 25s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/65ae4727a2da4e2eb691775289671d09 : SUCCESS in 2h 51m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0610cea023144869852748f16818cb43 : SUCCESS in 2h 01m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f908a7aca39f48a5b4d4de37239ba16e : SUCCESS in 1h 28m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab32ef8be28647678a298467e35cca8b : SUCCESS in 1h 10m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1e959df6a8341ad90e10ba78e0cc8c7 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"75bbb780bcddf1cceda2ac5c8a65f74f317a3458","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-19 12:55:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3c4def43994fc9b286d692b99a1fc29d5bc3bf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 12:55:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b91c0daf55929f02ad299ed95e977a3dfd5838bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:22:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/98d4914ff0224560989fdbe769b7400a : SUCCESS in 12m 13s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/37db1dd4d9ca4bde968de2c27844d358 : SUCCESS in 1h 34m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1a8a4c19912d4a0997e1beac7e58898c : SUCCESS in 1h 45m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcee007ba3dd46deb2d3bbfc92bdcb2f : SUCCESS in 1h 41m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50ebe6fddef44a09a3a3b3f47b7d549f : FAILURE in 2h 15m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4313280deded4bcca1dacaad4a8b1431 : SUCCESS in 1h 23m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/789b252b8c78401cad2e83d6fe0410b9 : SUCCESS in 1h 12m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d8b1e87fb449428b8332e43e6ccbc2 : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a1b9460fbf81d299704679fba4b2127da689ed2c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-19 15:37:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd700862c21edb27759df3c2bede67531d56929f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 18:22:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/82ec4b045c534d2e9241b2857db9a376 : SUCCESS in 10m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dd8068c375c5432286a568fd27cc8920 : SUCCESS in 1h 38m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/11fa86bac5534b1b98d52c9baec148d2 : SUCCESS in 1h 48m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7a9c72ad3254b95b52a668d118e5ee8 : SUCCESS in 1h 34m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/094bd055e04748f48a69e5aa5c97d3b9 : SUCCESS in 1h 25m 55s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bcef4d14c5f2451c811cdb79ca755b12 : SUCCESS in 2h 33m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ef1d2702c88483c892e450a88944293 : SUCCESS in 2h 02m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3282aca908894c63abd5af737c039369 : SUCCESS in 1h 17m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f2020d7a52741eaa57c31ac650e30ac : SUCCESS in 1h 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc935b788004eb6888b5d4f649881c8 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8f81f6d62ed58e840ba9af4968fd4597b70cfda6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 18:22:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3d0204ebb92dba8145fd5dbf24e3fc6f0c74e32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 20:25:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5d37abf84885efeed3e42cc83607e9327fceacb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 20:25:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4fee02cb3e25400ea70649e94d400826 : SUCCESS in 15m 19s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/30ce237c152240d2a0b2c321c6892c5a : SUCCESS in 1h 31m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bf56eea94a8b4f14a43501933433bda1 : SUCCESS in 1h 48m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fb18259437c45e8b478a723cea54bb6 : SUCCESS in 1h 47m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/271e6aa2a3d34e71914a0472b9a93ff3 : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/21eb856d28f74b6ebddfb8529830ec17 : SUCCESS in 1h 22m 39s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12fb08b8559f44248b47d631b9919d53 : SUCCESS in 1h 08m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a649cf0e5674e349faa4fee6e206e8e : SUCCESS in 9m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e2a3c167aa2cc894caaaedc8a53008f99608be02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 20:27:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e629e04381cd40409498d99075c15851 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d015850a361a8f5e93eb9c8184046bdae1f393d","revisions":{"0d015850a361a8f5e93eb9c8184046bdae1f393d":{"kind":"REWORK","_number":1,"created":"2020-10-19 09:19:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/46/758746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/46/758746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/758746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/758746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/758746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/46/758746/1"}}},"commit":{"parents":[{"commit":"c2f9675e0f88b62b893f285f0eb3a0f0f5e0808e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/c2f9675e0f88b62b893f285f0eb3a0f0f5e0808e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 09:19:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-10-19 09:19:57.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I8b8102da999ba06d1ab7039db029390726027380\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/0d015850a361a8f5e93eb9c8184046bdae1f393d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/0d015850a361a8f5e93eb9c8184046bdae1f393d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
