)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~764170","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I87eec21bdd620d65a41793d288c6eb25a14f0825","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","hashtags":[],"change_id":"I87eec21bdd620d65a41793d288c6eb25a14f0825","subject":"Dep\u0027s should be restricted by tox-constraints","status":"ABANDONED","created":"2020-11-25 14:01:15.000000000","updated":"2020-11-27 01:07:23.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c65500e849548241ec19929e7709a1ae1e7c0962","_number":764170,"virtual_id_number":764170,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-25 14:14:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"all":[{"value":-1,"date":"2020-11-26 13:45:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 14:14:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-26 13:45:11.000000000","updated_by":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"reviewer":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"state":"REVIEWER"}],"messages":[{"id":"aeca8143dee2d79b50b32092695c7a7d0bcfad5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-25 14:01:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a7682c10a7f799321e67b6fd8cb3e50ee872937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:14:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4ced7a8af9534e0999dad5a1ff650676 : SUCCESS in 9m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3ce987c4514712a69d005955fa86b1 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5d9976a86514686bd9807097590457e : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"edc97fbeb48b5163beec2740e4794745f46f69b1","author":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"date":"2020-11-26 13:45:11.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"c65500e849548241ec19929e7709a1ae1e7c0962","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-27 01:07:23.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5823d3e9f3b064548dbfd670933f99c7154a36f","revisions":{"c5823d3e9f3b064548dbfd670933f99c7154a36f":{"kind":"REWORK","_number":1,"created":"2020-11-25 14:01:15.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/70/764170/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/70/764170/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/70/764170/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/70/764170/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/70/764170/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/70/764170/1"}}},"commit":{"parents":[{"commit":"b73d93efd095b94f3983ff7f3a057deadcededb3","subject":"Merge \"Remove duplicated gnocchi_db_setup_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b73d93efd095b94f3983ff7f3a057deadcededb3"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-24 15:36:35.000000000","tz":0},"committer":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-25 13:59:31.000000000","tz":0},"subject":"Dep\u0027s should be restricted by tox-constraints","message":"Dep\u0027s should be restricted by tox-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s tox-constraints.txt\n\nChange-Id: I87eec21bdd620d65a41793d288c6eb25a14f0825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/c5823d3e9f3b064548dbfd670933f99c7154a36f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/c5823d3e9f3b064548dbfd670933f99c7154a36f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
