)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~768659","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Ia77d0b485dbcae7ac53c3dcb7e37c3ddac066906","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","hashtags":[],"change_id":"Ia77d0b485dbcae7ac53c3dcb7e37c3ddac066906","subject":"Add Victoria release info","status":"ABANDONED","created":"2020-12-29 03:12:27.000000000","updated":"2021-01-26 17:30:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a3b44e3cb7a646045d7101e1ff4cd490ecb91056","_number":768659,"virtual_id_number":768659,"owner":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-29 03:20:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 03:20:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03aeac4f62a256e17d59348833508f2487a54f8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"date":"2020-12-29 03:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97028cadcca8a8dec1fbbf922afe043e9dbbadae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 03:20:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/78c9d59f7fe245c8ac597ba647e38744 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640f2c832cf14056a826d7eae86c03d9 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55db602da55d4e058e1d15fb7fd44022 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a3b44e3cb7a646045d7101e1ff4cd490ecb91056","tag":"autogenerated:gerrit:abandon","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 17:30:16.000000000","message":"Abandoned\n\nWe are no longer publishing releasenotes for this repo","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d69e7e3328f3ca360d34a8a22862bce0da2da92","revisions":{"4d69e7e3328f3ca360d34a8a22862bce0da2da92":{"kind":"REWORK","_number":1,"created":"2020-12-29 03:12:27.000000000","uploader":{"_account_id":31828,"name":"yang yawei","email":"yangyawei@inspur.com","username":"yangyawei"},"ref":"refs/changes/59/768659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/59/768659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/59/768659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/59/768659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/59/768659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/59/768659/1"}}},"commit":{"parents":[{"commit":"b73d93efd095b94f3983ff7f3a057deadcededb3","subject":"Merge \"Remove duplicated gnocchi_db_setup_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b73d93efd095b94f3983ff7f3a057deadcededb3"}]}],"author":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 03:11:18.000000000","tz":480},"committer":{"name":"yangyawei","email":"yangyawei@inspur.com","date":"2020-12-29 03:12:19.000000000","tz":480},"subject":"Add Victoria release info","message":"Add Victoria release info\n\nAdd the lack of release information for Victoria, this patch added it.\n\nChange-Id: Ia77d0b485dbcae7ac53c3dcb7e37c3ddac066906\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4d69e7e3328f3ca360d34a8a22862bce0da2da92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4d69e7e3328f3ca360d34a8a22862bce0da2da92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
