)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~772281","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I5282cc78db55dee5f43cc34486a71a69567f2a28","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa-new-pip","hashtags":[],"change_id":"I5282cc78db55dee5f43cc34486a71a69567f2a28","subject":"Move gnocchi pip packages from constraints to requirements","status":"MERGED","created":"2021-01-25 10:02:57.000000000","updated":"2021-03-04 10:11:17.000000000","submitted":"2021-03-04 10:08:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"772281-osa-new-pip","meta_rev_id":"5a3a4ed6ab59eeacd0a87472cddac6963e9cc5ca","_number":772281,"virtual_id_number":772281,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 10:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-03-04 10:11:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-03-04 10:11:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-03-04 08:43:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 13:14:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-28 16:42:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2021-02-08 10:07:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2021-02-12 10:05:22.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"353b0661519a55e97f1ab6a379cba8b18bbce51d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-25 10:02:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc83863129cb8f3758cece41e9b7c13f02cc837c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 13:14:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/76df60fc9a414c8fa175ba554f9115dd : SUCCESS in 10m 57s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/17b2730395f2427684591db409f3c06b : FAILURE in 58m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/db776837fcd547aaa1505b743b1eab12 : FAILURE in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fbf495187d24648935d54bc8a3ee4d5 : FAILURE in 1h 18m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26a422a364114ddc9f41836c92f1b095 : FAILURE in 1h 05m 28s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9985db6e42c74c91b4a7efce328b6023 : FAILURE in 1h 43m 34s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/056c7fd7d0e54bd6be1f319c8f35e141 : TIMED_OUT in 3h 01m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9f1961215ddb4d92af87e0c19a61c71d : FAILURE in 53m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67bdb267004841099f409c5706b4e6db : FAILURE in 48m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c8b6b8c91114d2fb9773fe2591bf5af : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"486c47bcc79d9a2cc12f1f422b5ca71186a988b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 09:12:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c2eacfc588610324bb8dac422178aaf7783d708c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 12:40:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d4588ca923c84138b7b5c49b7a072740 : SUCCESS in 8m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/cc08d30d293d4265999e1539bb024b61 : TIMED_OUT in 3h 04m 16s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f3b58a99085b40279c1c09351890624d : TIMED_OUT in 3h 05m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c29b587cdd049e49e45f0054e6e9db5 : TIMED_OUT in 3h 05m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b37afc7b78654fa99a34060dcddd3d62 : TIMED_OUT in 3h 04m 20s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a6faf8f3dff44e8198acae12bbdc06ce : TIMED_OUT in 3h 06m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a17f98de1c740af8fadfc94e6cce401 : TIMED_OUT in 3h 03m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/086c1d2ab0f74ab8a62f43c7c1d92359 : TIMED_OUT in 3h 04m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcee35a6c823403baf238ffde8acbbbd : FAILURE in 22m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05179d13ae1c4e8f8eaad0e572bdcc23 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5fa583765ed8539c5d4e4e719f01ea7ee8276918","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-28 16:42:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"96586b6801bebd1693cc5adc923125d28af44a8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-28 16:56:40.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"8be0772e838df01b9afa1759a18fe38554db5072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 20:33:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1fae1303d4774a058e6eee6ee1cb2012 : SUCCESS in 8m 39s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d3e5e47573ce442688b75d6f9b8032b0 : TIMED_OUT in 3h 04m 22s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f9ce7b3c4aea4326873260e0caca4619 : TIMED_OUT in 3h 05m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96789a13b5864123acda9a630bdc478b : TIMED_OUT in 3h 04m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e874853dbe1546ac9fc149cef8506985 : TIMED_OUT in 3h 04m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/28e8a45a33d947b6a41677348bdb1a3e : TIMED_OUT in 3h 07m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/446abf69ac6e44a38d639569e6a4b677 : TIMED_OUT in 3h 02m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bdf499c149b24816b674205104ae1ce3 : TIMED_OUT in 3h 02m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f36c60e9bd744e9886fb987798084fa3 : TIMED_OUT in 3h 07m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a25cab7fd1884fdfa808e9e050d2ff31 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7ef9427e5f83eb27a26bf6b75f27151ad4752954","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-28 22:20:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"b1d6f7f43021eb0de01526a2cdd4db10534add32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 01:39:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4a5ffae5dfb448f5a5856d060f916759 : SUCCESS in 12m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/04e95361a1944d0ca4ee58f6101899f7 : TIMED_OUT in 3h 04m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8552e5014d2244a8b9fd9ce898dcf3f6 : TIMED_OUT in 3h 05m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd32b6828c0c4d918bdc8e74b6af3848 : TIMED_OUT in 3h 05m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d19cafcb79c3480582b3cd5d47180544 : TIMED_OUT in 3h 04m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bd56501f890041c2907c8a225fa76dc5 : FAILURE in 18m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee8ba790cd144b259f12073a752332a2 : FAILURE in 10m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/be9a962fe7504633bb018230d60de340 : TIMED_OUT in 3h 03m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fefb4aadb7d7461faae08fce87f051fe : TIMED_OUT in 3h 03m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa6a9d38f5d430b85762468e08d046f : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"ab349c8c54a1d92e5ddfeb9da7f3f697395c0727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-29 06:53:21.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"ba8a2d4c6ad8361117396b35b63cc47d817d9798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 10:16:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f515ee9de43f495b8f5fc93cb74907ca : SUCCESS in 8m 37s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b4d37459a03c478d8eaace26a07aac0c : TIMED_OUT in 3h 04m 56s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/82e0044ffe184795b634f6e60359c124 : TIMED_OUT in 3h 05m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b340d86176c742b88f375c3da53739a1 : TIMED_OUT in 3h 04m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ab0c6939ff24201a86b6f8c3c7f7b39 : TIMED_OUT in 3h 07m 15s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/69084872a8ca4e6e8d1df410ac559a9c : FAILURE in 15m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d36ee8c551e343238e89ec1f991c0cf7 : FAILURE in 9m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2ad692afcea44640ad95c7230f1c95a8 : TIMED_OUT in 3h 03m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df20f6ca84574eeb950061f7a97efc72 : TIMED_OUT in 3h 04m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54de2c4dcba640e88e3bc4fa3abe62cf : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":5},{"id":"147e2cd1486ed2eeb928f6d9a89361c40734e076","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-29 18:21:32.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"205fae80ada8897dae44cb22259530aa85a0aea6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-01-29 19:13:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dd99fe93303110cff66a97bac02b4987d9848844","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 21:34:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/206b7ed9d4914e728c73cc0e46d8343f : SUCCESS in 8m 59s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d77179d11a644e2e83eeaae37afafa89 : SUCCESS in 1h 20m 45s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d8c1388d762943ec999ef955448d4e65 : SUCCESS in 1h 47m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a9d94f0ed744dcfb82750debca530e8 : SUCCESS in 1h 55m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cd8b93a18bf4256b595f56d9e63fe9e : SUCCESS in 1h 25m 53s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/affe7e26d25f42508504a2ff7a9b31c7 : FAILURE in 21m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8013ac597de142bf9ec20bfeebf9ed1e : FAILURE in 12m 10s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8e94d2feb03d46fb986ab21585f3f155 : TIMED_OUT in 3h 03m 14s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2783fd6e857c41d2a318c492c38e9601 : SUCCESS in 1h 03m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1caefa2c22314c84894d42fe505217b3 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":6},{"id":"5ac342346222f980384546eda19e283a0a975ec5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-05 17:20:40.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a2979d64b7bb7014a0dd7d842231ead9a86ca5fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 20:36:03.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6856b70582c1410e89f79eda5faf7d40 : SUCCESS in 12m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c742ed8fb65847c48b522b5b9fd9f3c9 : SUCCESS in 1h 40m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c68dfe8f5a7346c1b6bcb3d40433dfa1 : SUCCESS in 1h 37m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c10cb8d9e664a829d311389fc538ae3 : SUCCESS in 1h 39m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81b3ad896f4d4c5fb1654101d5d3e4ba : SUCCESS in 1h 19m 59s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/212fea090525410e8da0daaf9ae38b5d : SUCCESS in 2h 23m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1a639b8fec4494b8d2e9cc7937dfdcf : FAILURE in 1h 16m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ca16a7588d864e0ca8e4e236a45b2739 : TIMED_OUT in 3h 02m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/490a4d4ac48e4985b7679a00a01bad8c : SUCCESS in 1h 00m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa6c9da59f92481984e6b6f144290955 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":6},{"id":"ddfc8ed2c512125510ea571cfc857629fc85c1a0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-07 19:39:23.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"a201473df6716f9a519fd04d3a4c4a96a7e213ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-07 22:46:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/edff032e300b44ddb1f69d89e5b95441 : SUCCESS in 8m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/dbab4bb80bcc4f49929bdd0ba3fc64bb : SUCCESS in 1h 27m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fe306d464a83426189b7f53cd091c7f3 : SUCCESS in 1h 44m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29c4da938e8b4f32ade959eee7c03c52 : SUCCESS in 1h 25m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c57200cafbb4707adcfc7384144bde6 : SUCCESS in 1h 10m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/810afa0a2e6e410f843698ad76af59f2 : SUCCESS in 2h 19m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6aef0af919d2403b872593cd1c1b7ef8 : SUCCESS in 1h 52m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/29b7863c729c4d4ab9710a36156a0724 : TIMED_OUT in 3h 02m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0195f8e5337446d096340164d463a84b : SUCCESS in 58m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89278b434107436f9af6c0ffc00c83c1 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":6},{"id":"c3918f3b2407c1b48987f4746f4d65907e255ee7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-08 10:07:26.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f18229763d2321071e55b4efaa6b0d5e4d3894c9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-12 10:05:22.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"540c8183cef4bd22187640b87b27af4a26aaae11","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-02 07:48:03.000000000","message":"Patch Set 6: Workflow+1\n\npush to gate","accounts_in_message":[],"_revision_number":6},{"id":"b150658fd5b3656d758a179381db5e0bb41df06b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 07:48:26.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b9ebd4eec3ec2d97df82dc606f38aac1eee8504c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 09:22:21.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3ec6746549fa405f879b2543e22dcdfc : SUCCESS in 8m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bc665df9f48d4eb8b87dccd445696099 : SUCCESS in 1h 18m 07s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/25f31193f7324856afc7dbac5ac2e1a5 : SUCCESS in 1h 28m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c56365c9a754daab8a51bd03e31be95 : SUCCESS in 1h 30m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/945d4bb6005e4f94bf96309b856be451 : SUCCESS in 1h 15m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0dee06b6095e4a2cb5e21c367090cb95 : FAILURE in 37m 31s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90fc7c110eb94669b28d67c16fea8dd5 : FAILURE in 47m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c4c38fbe404d898a533780fa8c0225 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":6},{"id":"5ea9f26f122150363aa699637fb4b3160dbc11bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-02 23:25:22.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"afb2df1019f569b3a8c534be9ac593c7ceafdad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 01:14:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a4cdacb21ce14c259e41831c22f1db33 : SUCCESS in 8m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8b00297cf0694858a9ad6787fc758b04 : SUCCESS in 1h 08m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/47c6004fb9224b1894c002315f177342 : SUCCESS in 1h 08m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61a07caab78f4432a06b92ec2b1b31e8 : SUCCESS in 1h 15m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c86b9d9d5304cd09231a35f84d4abc4 : SUCCESS in 1h 16m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/baf2e77f72644c15ab0139cdc05399b7 : SUCCESS in 1h 45m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f84722b1b4a14239a1176c4cc484933d : SUCCESS in 1h 34m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cbb5e6cd615f4a8da1b4b329d2560b89 : FAILURE in 32m 35s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1e08f0437164652877f11358477e6dd : FAILURE in 42m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb7d954ea06485ba0b64dc287a956c8 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":6},{"id":"48a1bc7cf5bb72538ab9fb35f0df83ee0d32de0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-03 10:10:20.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"38a0e5818748469bfa94cbd4970f136dbff75ca3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 13:25:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/cb0368be719b462a819cf72018df1d3d : SUCCESS in 15m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0f83a92e450349678dfd5ce487c8e63b : SUCCESS in 1h 26m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ffccc3b10fca4d2d9eb156c836fe9548 : SUCCESS in 1h 28m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54064d18ea2543778083f187228c0577 : SUCCESS in 1h 50m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30e2959d18a143faa906e06ea0d2eedf : SUCCESS in 1h 29m 51s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c5d856beb4954533a63fceb621588d42 : TIMED_OUT in 3h 06m 08s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/127b9d84b5ce4a9a9b78fbfce3737ea6 : SUCCESS in 2h 14m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e82e03f17304043947a8ef39b29060e : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":7},{"id":"971e55fae24c8b13f324889b3c1159fda0d58ade","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-03-04 08:43:30.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"f3328892c21429ed43edb36b34174e5551eab123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 08:43:46.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2d81216c14bf8e68308ef6f15ee7d861e7797480","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:08:23.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0b93cd5cb6e74955a091529fa9b015ac : SUCCESS in 10m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b71a1ebc57a04bb99ac6cb71cecaaa39 : SUCCESS in 1h 09m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f5a118cae3f74accb44c967dd221e92a : SUCCESS in 1h 15m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1f26778f84542f8be15c1e8225e8a3b : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6ed44c4cbd24971b856679fe4e3e823 : SUCCESS in 1h 04m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401a38cbc4ec46028c5931b51034a768 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":7},{"id":"e58cf534f73b9393e613e97d5997314f1f633b05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:08:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"4029bedc9eff5960e93dcbc33e7e330bce17b99a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:11:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/647fec8ffb3947de9396070b9c5f7961 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a51416662ca737262576319d8eb2a846d7e7639a","revisions":{"debadb0c02e1b6a51f44efbcb4791c8ec5afff89":{"kind":"REWORK","_number":1,"created":"2021-01-25 10:02:57.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/81/772281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/81/772281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/1"}}},"commit":{"parents":[{"commit":"cfa854fd067685d28ef51adc5a2c57514b9f354f","subject":"Use global service variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cfa854fd067685d28ef51adc5a2c57514b9f354f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:02:07.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:02:07.000000000","tz":0},"subject":"Move gnocchi pip packages from constraints to requirements","message":"Move gnocchi pip packages from constraints to requirements\n\nThis is necessary for the new pip resolver\n\nChange-Id: I5282cc78db55dee5f43cc34486a71a69567f2a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/debadb0c02e1b6a51f44efbcb4791c8ec5afff89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/debadb0c02e1b6a51f44efbcb4791c8ec5afff89"}]},"branch":"refs/heads/master"},"1e7389163431160ed50b2c1704ec98468d741db1":{"kind":"REWORK","_number":2,"created":"2021-01-26 09:12:12.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/81/772281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/81/772281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/2"}}},"commit":{"parents":[{"commit":"cfa854fd067685d28ef51adc5a2c57514b9f354f","subject":"Use global service variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cfa854fd067685d28ef51adc5a2c57514b9f354f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:02:07.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-26 09:12:09.000000000","tz":0},"subject":"Move gnocchi pip packages from constraints to requirements","message":"Move gnocchi pip packages from constraints to requirements\n\nThis is necessary for the new pip resolver\n\nChange-Id: I5282cc78db55dee5f43cc34486a71a69567f2a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1e7389163431160ed50b2c1704ec98468d741db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1e7389163431160ed50b2c1704ec98468d741db1"}]},"branch":"refs/heads/master"},"715882fbf2578ab166d947fc6ebcf3046afa2e68":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-28 16:56:40.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/81/772281/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/81/772281/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/3"}}},"commit":{"parents":[{"commit":"cfa854fd067685d28ef51adc5a2c57514b9f354f","subject":"Use global service variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cfa854fd067685d28ef51adc5a2c57514b9f354f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:02:07.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-28 16:56:40.000000000","tz":0},"subject":"Move gnocchi pip packages from constraints to requirements","message":"Move gnocchi pip packages from constraints to requirements\n\nThis is necessary for the new pip resolver\n\nDepends-On: Idfdb34e53d7e0a54dd35b95e7b9f51eece405238\nChange-Id: I5282cc78db55dee5f43cc34486a71a69567f2a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/715882fbf2578ab166d947fc6ebcf3046afa2e68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/715882fbf2578ab166d947fc6ebcf3046afa2e68"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1ff1c70e46eb13ab3edbed60c582a3dcda3b3e6e":{"kind":"REWORK","_number":4,"created":"2021-01-28 22:20:54.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/81/772281/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/81/772281/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/4"}}},"commit":{"parents":[{"commit":"cfa854fd067685d28ef51adc5a2c57514b9f354f","subject":"Use global service variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cfa854fd067685d28ef51adc5a2c57514b9f354f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:02:07.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-28 22:20:47.000000000","tz":0},"subject":"Move gnocchi pip packages from constraints to requirements","message":"Move gnocchi pip packages from constraints to requirements\n\nThis is necessary for the new pip resolver\n\nDepends-On: Idfdb34e53d7e0a54dd35b95e7b9f51eece405238\nChange-Id: I5282cc78db55dee5f43cc34486a71a69567f2a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1ff1c70e46eb13ab3edbed60c582a3dcda3b3e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1ff1c70e46eb13ab3edbed60c582a3dcda3b3e6e"}]},"branch":"refs/heads/master"},"3a4cfdbb5fb5b8aaee515f3f33ac6e242f7cf5d9":{"kind":"REWORK","_number":5,"created":"2021-01-29 06:53:21.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/81/772281/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/81/772281/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/5"}}},"commit":{"parents":[{"commit":"cfa854fd067685d28ef51adc5a2c57514b9f354f","subject":"Use global service variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cfa854fd067685d28ef51adc5a2c57514b9f354f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:02:07.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-29 06:53:03.000000000","tz":0},"subject":"Move gnocchi pip packages from constraints to requirements","message":"Move gnocchi pip packages from constraints to requirements\n\nThis is necessary for the new pip resolver\n\nDepends-On: Idfdb34e53d7e0a54dd35b95e7b9f51eece405238\nChange-Id: I5282cc78db55dee5f43cc34486a71a69567f2a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/3a4cfdbb5fb5b8aaee515f3f33ac6e242f7cf5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/3a4cfdbb5fb5b8aaee515f3f33ac6e242f7cf5d9"}]},"branch":"refs/heads/master"},"276c70b9dafe213aa751628f56ef335abebc0d98":{"kind":"REWORK","_number":6,"created":"2021-01-29 18:21:32.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/81/772281/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/81/772281/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/6"}}},"commit":{"parents":[{"commit":"cfa854fd067685d28ef51adc5a2c57514b9f354f","subject":"Use global service variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cfa854fd067685d28ef51adc5a2c57514b9f354f"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:02:07.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-29 18:21:30.000000000","tz":0},"subject":"Move gnocchi pip packages from constraints to requirements","message":"Move gnocchi pip packages from constraints to requirements\n\nThis is necessary for the new pip resolver\n\nDepends-On: Idfdb34e53d7e0a54dd35b95e7b9f51eece405238\nChange-Id: I5282cc78db55dee5f43cc34486a71a69567f2a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/276c70b9dafe213aa751628f56ef335abebc0d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/276c70b9dafe213aa751628f56ef335abebc0d98"}]},"branch":"refs/heads/master"},"a51416662ca737262576319d8eb2a846d7e7639a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-03-03 10:10:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/81/772281/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/81/772281/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/81/772281/7"}}},"commit":{"parents":[{"commit":"503b4d25cf0ef9e28b6f35d0b2569ebbe5fca729","subject":"Delete gnocchi_identity_setup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/503b4d25cf0ef9e28b6f35d0b2569ebbe5fca729"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:02:07.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-03 10:10:20.000000000","tz":0},"subject":"Move gnocchi pip packages from constraints to requirements","message":"Move gnocchi pip packages from constraints to requirements\n\nThis is necessary for the new pip resolver\n\nDepends-On: Idfdb34e53d7e0a54dd35b95e7b9f51eece405238\nChange-Id: I5282cc78db55dee5f43cc34486a71a69567f2a28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/a51416662ca737262576319d8eb2a846d7e7639a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/a51416662ca737262576319d8eb2a846d7e7639a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
