)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~822860","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I451dbcda0d752cc7cdeaeae44f4ea9832d6c7bfc","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","attention_set":{"20705":{"account":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"last_update":"2021-12-24 07:44:28.000000000","reason":"Dmitriy Rabotyagov replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I451dbcda0d752cc7cdeaeae44f4ea9832d6c7bfc","subject":"add redis package to gnocchi.","status":"NEW","created":"2021-12-23 22:17:14.000000000","updated":"2021-12-24 07:44:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a09e9886d951b34b2015d76de793b1b74bf5350a","_number":822860,"virtual_id_number":822860,"owner":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-24 01:09:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":-1,"date":"2021-12-24 07:44:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-24 01:09:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-24 07:44:28.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"a007e873308ab119316b71daab507136c0fec2ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"date":"2021-12-23 22:17:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1654acb562ab16bd4f98ce4e2edf277fd333947e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"date":"2021-12-23 22:21:26.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d972d865d2f2b79a1e3dfa8b6f20f671b91f2d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 01:09:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60ba66258e664221b2b2f57089fb6439 : SUCCESS in 17m 12s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/172ebab6a392433c8638c810be5e8c8a : SUCCESS in 1h 10m 57s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/bcb37927e0cd470ab81d3a6667f0b433 : SUCCESS in 1h 19m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/7619cb7e880540e18feed6375933be3e : SUCCESS in 1h 21m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b8e43a9928742d5acfc7e5e9fe0d349 : SUCCESS in 1h 21m 02s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f013b391de7f4264b654570dd3897c5c : SUCCESS in 2h 24m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e67b213cb42642508d121bc962eb25cd : SUCCESS in 2h 08m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d0b1ddba38546d09c0002b14d2aadd2 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":2},{"id":"a09e9886d951b34b2015d76de793b1b74bf5350a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 07:44:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0885692070afa9adf04500978893b3791cbdd2ff","revisions":{"02af386b827f1e440569efd59d19a4f6906645bf":{"kind":"REWORK","_number":1,"created":"2021-12-23 22:17:14.000000000","uploader":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"ref":"refs/changes/60/822860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/60/822860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/60/822860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/60/822860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/60/822860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/60/822860/1"}}},"commit":{"parents":[{"commit":"60643fa21c25c1f23d8765e72abaa874ecbc4e54","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/60643fa21c25c1f23d8765e72abaa874ecbc4e54"}]}],"author":{"name":"Shashi Dahal","email":"shashi.eu@gmail.com","date":"2021-12-23 22:09:26.000000000","tz":60},"committer":{"name":"Shashi Dahal","email":"shashi.eu@gmail.com","date":"2021-12-23 22:16:47.000000000","tz":60},"subject":"when incoming driver used is redis - which is typical in big/public cloud deployments, without this package, the installation is stuck on db sync operation and never times out","message":"when incoming driver used is redis - which is typical in big/public\ncloud deployments, without this package, the installation is stuck\non db sync operation and never times out\n\nCloses-Bug: 1955676\nChange-Id: I451dbcda0d752cc7cdeaeae44f4ea9832d6c7bfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/02af386b827f1e440569efd59d19a4f6906645bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/02af386b827f1e440569efd59d19a4f6906645bf"}]},"branch":"refs/heads/master"},"0885692070afa9adf04500978893b3791cbdd2ff":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-12-23 22:21:26.000000000","uploader":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"ref":"refs/changes/60/822860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/60/822860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/60/822860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/60/822860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/60/822860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/60/822860/2"}}},"commit":{"parents":[{"commit":"60643fa21c25c1f23d8765e72abaa874ecbc4e54","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/60643fa21c25c1f23d8765e72abaa874ecbc4e54"}]}],"author":{"name":"Shashi Dahal","email":"shashi.eu@gmail.com","date":"2021-12-23 22:09:26.000000000","tz":60},"committer":{"name":"Sashi Dahal","email":"shashi.eu@gmail.com","date":"2021-12-23 22:21:26.000000000","tz":0},"subject":"add redis package to gnocchi.","message":"add redis package to gnocchi.\n\nwhen incoming driver used is redis - which is typical in big/public\ncloud deployments, without this package, the installation is stuck\non db sync operation and never times out\n\nCloses-Bug: 1955676\nChange-Id: I451dbcda0d752cc7cdeaeae44f4ea9832d6c7bfc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/0885692070afa9adf04500978893b3791cbdd2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/0885692070afa9adf04500978893b3791cbdd2ff"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
