)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~822905","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~If1442abd7978d44def2a4386f9d159b42c4939e3","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"bug/1955676","attention_set":{},"removed_from_attention_set":{"20705":{"account":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"last_update":"2022-03-14 13:01:58.000000000","reason":"admin1 replied on the change"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-03-16 15:22:53.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-03-15 10:41:39.000000000","reason":"Jonathan Rosser replied on the change"}},"hashtags":[],"change_id":"If1442abd7978d44def2a4386f9d159b42c4939e3","subject":"Add availability to define gnocchi_incoming_driver","status":"MERGED","created":"2021-12-24 11:41:59.000000000","updated":"2022-03-16 15:24:23.000000000","submitted":"2022-03-16 15:22:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"822905-bug/1955676","meta_rev_id":"83d25613b03c7ae9e57751b54eb0b707971d7297","_number":822905,"virtual_id_number":822905,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-16 15:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"recommended":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"all":[{"value":1,"date":"2022-03-14 13:01:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"value":2,"date":"2022-03-16 13:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-03-15 10:41:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":0,"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"value":1,"date":"2022-03-16 13:32:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-24 14:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-04 14:15:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-01-04 14:15:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"state":"REVIEWER"},{"updated":"2022-03-15 10:41:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-03-16 13:32:45.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"21ae73f4d55bb61e4029971eadc4a6af67e933cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-12-24 11:41:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8aba1aad66dbdb0c009c49e6eeabb343217170d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-24 14:49:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a97d8a4003d4be094c0bd91dfdff80b : SUCCESS in 19m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9a015520c9e04c4f8c31f6f618f48926 : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2e65cc6fe9fb4f1a8ec1855d013adad1 : SUCCESS in 1h 13m 20s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/856bde43294140cab8142e3bb57d9fce : SUCCESS in 1h 00m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32026c06040845e49d199d6ac699b625 : SUCCESS in 1h 32m 39s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2f482168a044db9a6acbd20bd8fc25f : TIMED_OUT in 3h 04m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5d7e0f83d5945789a093cd03d01f1a5 : SUCCESS in 1h 52m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8afbc9ee3a7242dfb0b804dc1a098d35 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ae0452714664ff4924ed843a1353b4e : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"742bf9555f1538cb5a3ae7bde2b0b69d13c6a1fc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-04 14:15:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"808e0341a525655f57dfa0a2a1a08ffcebb14cc1","author":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"date":"2022-02-12 12:18:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df52687e9c10028f78ccbca430ae906d25476f11","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-03-03 15:38:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbe2f7719c4354502ab6d2e356521665a935a91e","author":{"_account_id":20705,"name":"Sashi Dahal","display_name":"admin1","email":"shashi.eu@gmail.com","username":"admin0"},"date":"2022-03-14 13:01:58.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26c5993ecfe8ca01ffb8a49217c54305f6d8f34d","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-03-15 10:41:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0f3a69f4d8732a9e0fde72e7a74d295ae19389b","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-03-16 13:32:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f33da81b8cb6d4f5f8514d07519c0839fff882f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 13:35:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5d91958d0c45bc76b0ea5906314e4e01362f119","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 15:22:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f04dde3affb7402bbbe9fc4ed64374fd : SUCCESS in 17m 56s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/35072ade9ddb4cecbcc817032bfd51c8 : SUCCESS in 1h 25m 13s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/363144766dd541f79b9da17efe45da36 : SUCCESS in 1h 26m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb280d63d5a445869099b93bbe81fc7b : SUCCESS in 1h 46m 53s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/cce651acc65c4f9f915c53f2e663a07c : SUCCESS in 1h 35m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11754ae2e28f488ea9e66589a467544c : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae80ea0042084753ae99d4599671e35b : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f95fd24166221d8d08bf60281cc6707199b20ea2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 15:22:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"83d25613b03c7ae9e57751b54eb0b707971d7297","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 15:24:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a5601a7d6e4fb9963ebca37d8c9bab : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84150e8fb17a0a3520358713b938da320a575500","revisions":{"84150e8fb17a0a3520358713b938da320a575500":{"kind":"REWORK","_number":1,"created":"2021-12-24 11:41:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/05/822905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/05/822905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/05/822905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/05/822905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/05/822905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/05/822905/1"}}},"commit":{"parents":[{"commit":"60643fa21c25c1f23d8765e72abaa874ecbc4e54","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/60643fa21c25c1f23d8765e72abaa874ecbc4e54"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 11:39:12.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-12-24 11:39:33.000000000","tz":120},"subject":"Add availability to define gnocchi_incoming_driver","message":"Add availability to define gnocchi_incoming_driver\n\nGnocchi supports having different storage and incoming drivers [1] which\nhas been never implemented in role.\n\nWe add `gnocchi_incoming_driver` variable and do incoming configuration only\nwhen it\u0027s not same as gnocchi_storage_driver.\n\n[1] https://gnocchi.osci.io/install.html#configuration-file\n\nDepends-On: https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/822901\nChange-Id: If1442abd7978d44def2a4386f9d159b42c4939e3\nRelated-Bug: #1955676\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/84150e8fb17a0a3520358713b938da320a575500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/84150e8fb17a0a3520358713b938da320a575500"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
