)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~890100","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Idbd2bef71983f6fb3cc402563a166349621bf03c","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa/db_port","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-22 17:10:48.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Idbd2bef71983f6fb3cc402563a166349621bf03c","subject":"Use proper galera port in configuration","status":"MERGED","created":"2023-07-31 13:08:33.000000000","updated":"2023-08-22 19:23:18.000000000","submitted":"2023-08-22 19:22:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890100-osa/db_port","meta_rev_id":"23eb7f2cfd7e5f36b91987d331c20176fec5d2f6","_number":890100,"virtual_id_number":890100,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 19:22:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-08-22 07:27:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-08-22 17:10:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2023-08-22 17:10:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-31 15:37:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-22 07:27:34.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"b93ab00ce8e8e497e56919cb37314b0c8e55578c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-31 13:08:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd3b9ff5e94640aab16247f321c9ab11762204f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 15:37:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98e56d74c5424d19ad0481893b5f4fbd\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1ae5eb99d694381a800ff3a70e8968d : SUCCESS in 13m 14s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f302df7507274584b42375c75248b01c : FAILURE in 1h 11m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/34c83859335d4736b5047f3073de8f33 : FAILURE in 1h 25m 20s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/303a1fbe330b4b7b86e6b6e29f89f503 : FAILURE in 1h 42m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cff54b41ba8242a18674c12f85d0fd4d : FAILURE in 26m 22s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a92dd4cc9fa4e028563cce187637f11 : FAILURE in 1h 35m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b34ea8b944f7467eaab8922788f35095 : FAILURE in 1h 33m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9755e25736d5486eaca698ea36115dda : FAILURE in 2h 25m 48s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f7bce4031744bc4b6e2ebc058b9fb63 : FAILURE in 24m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5d4fa6381845a0b01c068756895dc0 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"01cb199d2a10e75153ea3479e22573044fbf74bd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-07 07:03:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b055f10a599adf989f68a4452cc25736e50f77c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 09:42:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e833774d44e44f34bc4be88ad6266ff9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/202b4c2e341247b790e14564ca00c615 : SUCCESS in 11m 51s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/233a2cacc78444d2a62725d5776828c5 : FAILURE in 1h 16m 59s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6b2b97ad1ee4718bd922eac7826297c : FAILURE in 1h 14m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb189020c1484cb58784ab7a89039b96 : FAILURE in 1h 36m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c1e162677df4317ad424a3141751fc5 : FAILURE in 1h 12m 07s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/63367a66fc8943b784f75fca0247c5e3 : FAILURE in 1h 20m 05s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/580c90b43d16437684eb3328997466e4 : FAILURE in 1h 41m 20s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/78e6906ad555438fb8d4efd21907b69f : FAILURE in 2h 15m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72f2636ce14f44e6bc8add12e865d15a : FAILURE in 2h 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e38b7dde9714709b2d008b453123d40 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"7e54a92d9673d4362f9053d217469ebe19c2eee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-15 13:35:09.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"193d1ebc24d82c6efacfaa966b7f18115cd0798b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 20:19:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/836afac4bd0347068201d2ed82a609cf\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd97c013604146d881e159c9f54ef2c0 : SUCCESS in 10m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6d7bb82e45964d208e6ee2d6870e4554 : SUCCESS in 1h 50m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3405882e37dd44f59f635bb38cbf8b8a : SUCCESS in 2h 09m 50s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2ce8ad9a9719455e9a025ca20154fa4d : SUCCESS in 1h 50m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/babf31d9d7cf44949d8771086bb7e10d : SUCCESS in 1h 57m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ba0f581e184a4991bd8259ca3a8d2b53 : SUCCESS in 1h 59m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16eacf1a2b044323aebb57ca671f9bff : SUCCESS in 1h 56m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ed518b0faeb04115aed1d6543dd24433 : FAILURE in 2h 35m 53s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed831d9f4a0343ecb45d4e508c0fcbbe : FAILURE in 2h 17m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61bf81ba155d486e9da9caa0c66999c5 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"710a6d5997acb5ba000a2af3217c5b8641d26a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-16 13:33:43.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"93bc31d24062f9a5f3de6099efe0ff5c118dbf64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 16:47:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1a240ac2b7f446ca3068c888aab3f96\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50990c10948c490aa7374d01f367d585 : SUCCESS in 16m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/7e9ba8b038f748c9b25a73ae3f0c4517 : SUCCESS in 1h 29m 33s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ad11d548c5444609dd70d7ead1a1104 : SUCCESS in 1h 45m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f5b5806812f444fd9356f00f94bbbac6 : SUCCESS in 2h 06m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f17b7913abc4165bda2d9e06bb4ed5b : SUCCESS in 1h 44m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/af77175660bc441b8043c6f882bf5483 : SUCCESS in 2h 05m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f401bb511a424dbbba6981784cd2773d : SUCCESS in 1h 59m 38s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5e5ace713f504dbfb285a6f6da3f5887 : TIMED_OUT in 3h 04m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a50f96921d5a4b9bbb28cc381776313c : SUCCESS in 2h 59m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22543c3bffe64eec8025d18b23702c76 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":3},{"id":"c620ecea832878914f7663d4ce88bf602d803c54","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-17 15:21:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8b9bf25a98375ce2fe08a076930a2b38087265e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 20:40:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20619052d6634d368f966f090fcd62a2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/369c9cb4c73b4841a60da73bc1c7efb3 : SUCCESS in 18m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/483e16bef11f4e5eab3fc7c59b092b79 : SUCCESS in 1h 47m 39s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/648623c3e2c64fbe8d3475e8eb19ed19 : SUCCESS in 1h 52m 44s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1dac87e29b8e4c2aabaeb7427e4955df : SUCCESS in 1h 49m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fa637360625451f889d0452629070c6 : FAILURE in 1h 08m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9154842068c94967b58670bf978375e5 : SUCCESS in 1h 59m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2fb5e7688f647949609670b4bb2fd9c : SUCCESS in 2h 02m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1f40585d71cf4949bd9802990335ea87 : TIMED_OUT in 3h 05m 32s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7368e402dca54fe48096ad20859351af : SUCCESS in 2h 32m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad09ea500f4f490f8e26d17f908156ea : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b4229c367894318f6083e81175b730e6b140d98d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 08:06:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"064ac12f9a18097097bd06c68485145fad335a2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 10:50:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb7ad0838e14595b58b5ee028f40dab\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f35749931fab409a848b8c0ffc434804 : SUCCESS in 12m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d78a9f1bfbdd4942b08e4d8df3593064 : SUCCESS in 1h 38m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b252a13deaec4d4191c07060707dc59c : SUCCESS in 2h 00m 14s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b45e6e51e50f4b0dbce9c8ff30c79f9c : SUCCESS in 1h 58m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f931d331eab64d378db9f2a20960b113 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0caa6090c31141248c646b7c4abb8440 : SUCCESS in 2h 08m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d58daf79c7c942008a9b735aa1020457 : SUCCESS in 2h 15m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/254eaf60060b4ec492b2410db3e72212 : SUCCESS in 2h 25m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18ec5f27b2024d2eb9f4a96294be5194 : SUCCESS in 2h 42m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa7bef8e64754945af7a8d2009e754cb : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c6f69567ee0f66f611c9ea5291545fb8064abc35","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-08-22 07:27:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"971653868be31b1856e71e3d59cf5ff35c126c64","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 17:10:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6d4690303c20a679a71fd1bca214f5b216c0a33d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 17:12:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5ee2b3eae3621ea69e3b04686f58e8a5b2246308","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 19:22:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bb08b52855043879e5dc7d7708a86a6\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60d73a1a73b544039d62319fd6998da6 : SUCCESS in 12m 34s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6cc32d677d2f42b0ab8fc93f4503b42d : SUCCESS in 1h 20m 58s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11885350304942cca7255783e6f486f7 : SUCCESS in 1h 55m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7de8e1af899480c94ae2db2bf619b07 : SUCCESS in 1h 54m 57s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/47c1d4a4cd344dea95c4330d9857f0dc : SUCCESS in 2h 04m 48s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e22759955544de694456f891c3bd48c : SUCCESS in 2h 02m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ca11b22be344389952985585ef8689b : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3b501da8f0fb6e27f4da324ec1dfc11ce7a7c00e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 19:22:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"23eb7f2cfd7e5f36b91987d331c20176fec5d2f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 19:23:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1edf433f4980410a86ed28c4ef4a6fd9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef60c804201a449cbbff9835bafadd9b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2ec800116e53a36edcee23795f38dc6c699d5cc9","revisions":{"8d2451be50e98c92c5aef8797b07f76a05c656b4":{"kind":"REWORK","_number":1,"created":"2023-07-31 13:08:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/890100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/00/890100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/1"}}},"commit":{"parents":[{"commit":"e0e213efce5e195e93812d088b045d893495c771","subject":"Add TLS support to gnocchi backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e0e213efce5e195e93812d088b045d893495c771"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 13:08:24.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 13:08:24.000000000","tz":120},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservice\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nChange-Id: Idbd2bef71983f6fb3cc402563a166349621bf03c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/8d2451be50e98c92c5aef8797b07f76a05c656b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/8d2451be50e98c92c5aef8797b07f76a05c656b4"}]},"branch":"refs/heads/master"},"2b5836ae6f108fcf5e6d3cedbd4776541374b2d7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-15 13:35:09.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/890100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/00/890100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/2"}}},"commit":{"parents":[{"commit":"e0e213efce5e195e93812d088b045d893495c771","subject":"Add TLS support to gnocchi backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e0e213efce5e195e93812d088b045d893495c771"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 13:08:24.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-15 13:35:09.000000000","tz":0},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservice\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/891474\nChange-Id: Idbd2bef71983f6fb3cc402563a166349621bf03c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2b5836ae6f108fcf5e6d3cedbd4776541374b2d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2b5836ae6f108fcf5e6d3cedbd4776541374b2d7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2ec800116e53a36edcee23795f38dc6c699d5cc9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-08-16 13:33:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/00/890100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/00/890100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/00/890100/3"}}},"commit":{"parents":[{"commit":"e0e213efce5e195e93812d088b045d893495c771","subject":"Add TLS support to gnocchi backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e0e213efce5e195e93812d088b045d893495c771"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 13:08:24.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-16 13:33:42.000000000","tz":0},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservice\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/888517\nChange-Id: Idbd2bef71983f6fb3cc402563a166349621bf03c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2ec800116e53a36edcee23795f38dc6c699d5cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2ec800116e53a36edcee23795f38dc6c699d5cc9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
