)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~892356","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~stable%2F2023.1~Idbd2bef71983f6fb3cc402563a166349621bf03c","project":"openstack/openstack-ansible-os_gnocchi","branch":"stable/2023.1","topic":"cherrypick-osa/db_port-24yhx0eaik","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-10-27 12:57:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbd2bef71983f6fb3cc402563a166349621bf03c","subject":"Use proper galera port in configuration","status":"MERGED","created":"2023-08-23 07:26:07.000000000","updated":"2023-10-27 12:58:49.000000000","submitted":"2023-10-27 12:57:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892356-cherrypick-osa/db_port-24yhx0eaik","cherry_pick_of_change":890100,"cherry_pick_of_patch_set":3,"meta_rev_id":"98ca38081eaf0fc6f8a16cb0dbe82d4795c6c8c3","_number":892356,"virtual_id_number":892356,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-27 12:57:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-10-27 10:52:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-26 16:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-10-27 10:52:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 09:54:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-18 03:06:09.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"CC"},{"updated":"2023-10-26 13:05:32.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2023-10-26 16:56:53.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-10-27 10:52:11.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"7b594226c2822ad52a87dc97a4a45c7681ae0772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-23 07:26:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8a8c8a59ee0ccc69fb49510daf66471193965c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 09:54:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4dc16d107f2a4c72ade2cd55f849f78d\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce2d61582c1849619c9ef95933ac1b3f : SUCCESS in 12m 36s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e69c2fadfae644a5b81518cfb4072e06 : SUCCESS in 1h 35m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d71cacf9c1ab45efab30768add187d73 : SUCCESS in 1h 55m 57s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7d82c2a436d64968b8c52c6a674c1959 : SUCCESS in 2h 08m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e43001605b94df6a29b5d5b2673ae7f : SUCCESS in 1h 58m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abce42502d14481fb49c1c57b6e888d1 : SUCCESS in 1h 34m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/19edc045c36f404a86f4c8a171e0d9e0 : SUCCESS in 2h 14m 22s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2bd539134ec45bd970467f9cdb24282 : SUCCESS in 1h 55m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f8e5c81e4e794037b83a3b093448815a : FAILURE in 1h 12m 11s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07b19883bd3f4be9afe45114fb3af4dc : FAILURE in 1h 01m 44s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe1250877c164c1db7325c4cba3a5301 : FAILURE in 1h 08m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a99f8158ac4a24ac8fee6c2c88d9f4 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"06b195965e64c9942d10541146dbf72a61acbbf1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-06 16:08:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c206e87f445d8e1f5312985914ea57a76135c33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 19:24:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67b10b4f1f574681908b9e829b351214\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c1d4007a87543a08b903e6fea4694da : SUCCESS in 13m 02s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/cbb5288e08f646098bdb6940dd4bf35a : SUCCESS in 1h 46m 01s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/92c9874aebf94ea9942042a86e8267fd : SUCCESS in 2h 05m 06s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fd74e721582247b5b2a4ca6431fd8fd2 : SUCCESS in 2h 00m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e75d3512d18463f93ef537fe30c29ca : SUCCESS in 1h 54m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6a6a89239fd492fbdf3dbfb1dcf65e3 : SUCCESS in 1h 56m 08s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c1a9634049c14f7db4ba8127a58a0ae9 : SUCCESS in 2h 16m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f088f2a6f72746f4ba6a686ab67f2f5e : SUCCESS in 1h 55m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/73fdae0325204e17b435d0612da45463 : DISK_FULL in 3h 02m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d814be0892194f2eb4c7c32a54a4321d : SUCCESS in 2h 57m 13s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60003728e0774260b54b2f4aa60e2ae5 : FAILURE in 1h 05m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcd5bcb2ac94346a2aa368b088d38d2 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"42f1246a6b9fdc99f8b9406fc3fdf8c11585e41d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-10 17:20:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a77b54f8ab6c844508510fede215c401964f9eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 20:13:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f2e5d35e9514ec6a710e47f873f6483\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acc16459620f4fde881747f0535459b2 : SUCCESS in 14m 18s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/15e302bc046f4ce1b1540ad18242f167 : SUCCESS in 1h 33m 26s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/90601a0a64d642528eb676c2ad55d560 : SUCCESS in 1h 45m 10s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5983745dc28b4860a94f398cc6315cb4 : SUCCESS in 2h 05m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39bd9e22a45d454b97e063a8f6cdd039 : SUCCESS in 1h 49m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7be2319f63f417c9b65ff81eea35e52 : SUCCESS in 1h 58m 15s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8307c771be59433cba918775fd6a2090 : SUCCESS in 2h 20m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e787f60e5034a09a6746549a43047c5 : SUCCESS in 1h 41m 37s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eabb9395554141a78b738b786e9e79f2 : DISK_FULL in 2h 32m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2ca979675ed449688fa4b4f5bea827c : SUCCESS in 2h 36m 10s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ceb76d1449d43ac9846c416166a4156 : FAILURE in 1h 23m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c7a499bfd742e4a8599c3d2fa7bb59 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"326853d293ace3230a12a5a499e920a12dc25b50","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-11 10:15:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00381f2bc45064aff1367a7210dd3db69979cfc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 13:43:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05c6f9e9b4de46249f82ff7bbbaa1d12\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bff8c3c5be6045e4a70946d8d002ea0b : SUCCESS in 18m 03s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/46e8bb8d425647e0849e352b3950c820 : SUCCESS in 1h 49m 49s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e52ca555c2a4090ac196688f882e593 : SUCCESS in 2h 07m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a95252fdd40c4ab890c6f4689d7d0ffd : SUCCESS in 1h 55m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fe42649150d4ce1a930b875999907c4 : SUCCESS in 2h 04m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69c0b8c11e39423cafdbe8f04a77d712 : SUCCESS in 1h 51m 05s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0b9689667099497cbec4d5558fdb9b7b : SUCCESS in 2h 06m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7cdd9978da1474e853827dad632f05b : SUCCESS in 1h 40m 14s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/194e8131b24d4c54a2ef0e09bf0a3705 : DISK_FULL in 2h 57m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c52f1240112942acb927acaf280c30ad : SUCCESS in 2h 41m 35s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/278f696af3ca42c492fb032b34a76bf4 : FAILURE in 1h 00m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/755ba965d53f40ddbfcad6f3e95f40e2 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a85966e359daf7b035744649195bd8d2ddb8815d","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2023-10-18 03:06:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16c046b7799dee64048a9d08d05c43588fb7ad41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 06:05:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0604dd2ca5754f169fde5eb5564c7851\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9a1a3e71be74752b1e17794edd6f3b6 : SUCCESS in 12m 43s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/21f4a086d1814dc9bdd1277767a484f6 : SUCCESS in 1h 42m 08s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/807a47346219456b9b22be8563e287f7 : SUCCESS in 1h 56m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1168aa6fde8140d3b4c3d4de238a9156 : SUCCESS in 1h 39m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d62a5df14b3640fb8a3b92d0c8106f9d : SUCCESS in 1h 55m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b2873cc32b9485ca50f601de36aea4a : SUCCESS in 1h 36m 35s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9c64e9c813984837b1b039ed39747f26 : SUCCESS in 1h 55m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8c8b5659ee1491fb0f384d6e45c7d9f : SUCCESS in 1h 44m 22s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/64cd9d61998940729077a59699e402dd : DISK_FULL in 2h 14m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/360779fbf36f45e3a47a0b0343764ff8 : SUCCESS in 2h 50m 28s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c655fd65b0bc4854b0f400a744a389c6 : SUCCESS in 2h 29m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/215bf2c7e1104bf6bd6d264b7bafcdcf : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d832533b6dd36696eeee532d7d6c62447efe821d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-19 11:54:01.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"10a5069aae74aa9ef9208eca3c19834270ee1d16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 15:03:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cec08cc14452470e9cec670e3a3e1dd3\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d7c5824b732493a9b56010784eefb0c : SUCCESS in 13m 31s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/e5af3f0a7afb47e0a0d834f0b69104f9 : SUCCESS in 1h 21m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec19fc625afb46daa906149ffc24bf63 : SUCCESS in 1h 41m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/15a425ebda7540d5b8c6aec988a53d01 : SUCCESS in 2h 01m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a4fc9c5ac6c4d0babd9f47494a1a35f : SUCCESS in 1h 52m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17a59e2abad64c3290ef124e458e6886 : SUCCESS in 1h 37m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/44f56a96fbda47ec84aca7e5da3098d8 : SUCCESS in 2h 00m 06s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d65423b87f44e5da106e897ac74eb94 : SUCCESS in 1h 56m 07s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/77297b0a6b8c4a22b26ca4b1773c9007 : DISK_FULL in 2h 54m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afed479c8b42457db33bf2074bab4ea3 : SUCCESS in 2h 36m 09s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ab82918e0c947f2a84b926b56bf5241 : SUCCESS in 2h 00m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abf1ebf9780748f7b01beac93b4ece67 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"f44357f628da889f2b92bd6c38720497080dce8f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-20 23:10:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57dc2221c7f63b2324477fe052e66a9a0f76da66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-21 02:11:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b23b10a06482484eb189f0344795cc95\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36bfea70c2154d88b9dd202c8d9b0d67 : SUCCESS in 12m 19s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/a02b898743a7457fb016a584d3c9af56 : SUCCESS in 1h 38m 50s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/305136c8af3d46f7ad63b3d28b6246f1 : SUCCESS in 1h 45m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c5d2429d8b854a8ab52b2a0a0a513fd6 : SUCCESS in 1h 57m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96c8bee979654e04b0de1fe526c19a48 : SUCCESS in 1h 39m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e00f5ff28af143638065282c6392cfbe : SUCCESS in 1h 44m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d37a6afec83742e3b15947f77dab940a : SUCCESS in 2h 07m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/713978677b51466f917d65ab4072ecee : SUCCESS in 1h 30m 13s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8a32820e1061421c87c3a0b8651af676 : DISK_FULL in 2h 52m 14s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae792b831e084ee6936d94faffbcac65 : SUCCESS in 1h 49m 15s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b8bd94454714f0aa75248fe9002e583 : SUCCESS in 2h 20m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2966c83e184945b99ebbc4d56dd3e237 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b02c1bf7bb0e6cd6b3b2320c2777d30427be16bb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-26 13:05:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50d800d3e5a43345c6e5c021b8eeea4a7aa2009f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 16:18:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7397855cc7d54046b0481c7e8be80d76\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9840539d57a143728e155a3a0593b2bf : SUCCESS in 13m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/fadd6d34481540db9b5012c4d754e9df : SUCCESS in 1h 45m 22s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/cef29d89f1284c7599b9ffdf2d186bc5 : SUCCESS in 1h 56m 16s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e8468fdb6ec644cdb2fe2085a8e772b1 : SUCCESS in 1h 52m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3bb2a80e79e47b985ebaa213d6afb71 : SUCCESS in 1h 54m 11s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd0642bc3406449eac4ad68c87013bb6 : SUCCESS in 1h 36m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/638c92fa0d184e8e843d6626bf77dedb : SUCCESS in 1h 59m 27s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55643789cb484b84a04c454179be9425 : SUCCESS in 2h 01m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/eba991ea485d473faa781ebb43be3fd4 : SUCCESS in 3h 02m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/323bbb22e4e84b72942ae7e6a5186bb2 : SUCCESS in 2h 11m 31s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6317411a02ae408dac4fa1e2d533defb : SUCCESS in 2h 08m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f9a32038fee4c5199003fc5d87de628 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"84120eb71d1587e9d9e2456fed3d00bbe323fabf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-10-26 16:56:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a97ac5a56b3eef3815d949c53405dc8f42244579","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-10-27 10:52:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6691c7b81374e4417cdb6d70c7cb1880e228ebbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 10:54:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f5f73daef5a4e7e77dc5d1d930fbd6649c259642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 12:57:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360428333c3e48e1b5638052c50ca2ce\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03ee9804e51647c38d19f6dbaa1bed3e : SUCCESS in 12m 30s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/f0c33b05dd194a5fba3104d0be58b326 : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dd34c037f18e4951b16d3c311c19719f : SUCCESS in 1h 38m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc4ac45b59584daca8a00844da82c705 : SUCCESS in 1h 37m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5298ffef4e204bad84039fa5bb718f7f : SUCCESS in 1h 52m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab0edefe19484c3983d887688dd1651b : SUCCESS in 1h 53m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926403078e064f2baac786ec3ea9a492 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b1bf880ebcc383f404889b481a7c08aefecf246b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 12:57:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"98ca38081eaf0fc6f8a16cb0dbe82d4795c6c8c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 12:58:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4675f334126a46d182adaf3bafaabc87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a98bb5d40d451680f13654169df473 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c8be8efd9f61ff037e4c9e72bcd832c91d66bc44","revisions":{"2942453be8aa7d2bebc5622defe60a40a28410b5":{"kind":"REWORK","_number":1,"created":"2023-08-23 07:26:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/892356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/56/892356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/56/892356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/56/892356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/56/892356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/56/892356/1"}}},"commit":{"parents":[{"commit":"50849989ed3612ebdf10c263f1dfce734b8c2a31","subject":"Define constraints file for docs and renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/50849989ed3612ebdf10c263f1dfce734b8c2a31"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 13:08:24.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-23 07:26:07.000000000","tz":0},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservice\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/888517\nChange-Id: Idbd2bef71983f6fb3cc402563a166349621bf03c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2942453be8aa7d2bebc5622defe60a40a28410b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/2942453be8aa7d2bebc5622defe60a40a28410b5"}]},"branch":"refs/heads/stable/2023.1"},"c8be8efd9f61ff037e4c9e72bcd832c91d66bc44":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-19 11:54:01.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/56/892356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/56/892356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/56/892356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/56/892356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/56/892356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/56/892356/2"}}},"commit":{"parents":[{"commit":"50849989ed3612ebdf10c263f1dfce734b8c2a31","subject":"Define constraints file for docs and renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/50849989ed3612ebdf10c263f1dfce734b8c2a31"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-31 13:08:24.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-10-19 11:54:00.000000000","tz":0},"subject":"Use proper galera port in configuration","message":"Use proper galera port in configuration\n\nWhile \u003cservice\u003e_galera_port is defined and used for db_setup\nrole, it\u0027s not in fact used in a connection string for oslo.db.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/898770\nChange-Id: Idbd2bef71983f6fb3cc402563a166349621bf03c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/c8be8efd9f61ff037e4c9e72bcd832c91d66bc44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/c8be8efd9f61ff037e4c9e72bcd832c91d66bc44"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
