)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~894785","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~If05936d25db3329b4204172f3b800ddf3417b9b7","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"bugfix_change_policy_format","attention_set":{},"removed_from_attention_set":{"36356":{"account":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"last_update":"2023-11-28 09:08:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If05936d25db3329b4204172f3b800ddf3417b9b7","subject":"use policy.yaml instead of policy.json due to latest change on gnocchi","status":"ABANDONED","created":"2023-09-12 14:58:40.000000000","updated":"2023-11-28 09:08:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2aea23f882e698a249b21058f8a57662fd0af686","_number":894785,"virtual_id_number":894785,"owner":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-17 11:02:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 17:11:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"438aa416ce36738296e188855c60208206f712fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"date":"2023-09-12 14:58:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32cf6131816fa6915734a2c3f586dc1768dabafc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 17:11:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9bbc83c37e040c6b5792548db959da1\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/163d38bfd10d43158b4c9288aaa7105b : FAILURE in 13m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4664eb96552845d0b8697af9dfc66918 : FAILURE in 34m 23s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ea5d98c02f34bdd8160d87f73bc99f9 : FAILURE in 35m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e0983526e30b4bebaa4f0bf84a899280 : FAILURE in 42m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d52daab4ce624f019fbc2d8959566aa6 : FAILURE in 33m 34s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a06f479d8bc8439a80a0693bdc857b41 : FAILURE in 38m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d5201a2d267448986219b76c906424f : FAILURE in 31m 35s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/26043725bfff4b0ca0a82df3496d4f67 : FAILURE in 2h 05m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77a295f6265742428da32c33c97e142e : FAILURE in 1h 53m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4095dc7911ce4fe0a005455146d8ed77 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8307231769b9c66a09e21401c583595f3248288e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"date":"2023-09-13 09:33:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0d4f07fff5b55f1bc13e719eeaa1383261ef3153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 12:19:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07eedbb94bfb4cde8244ce072a9c7d71\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89c160b1d44647c19144ed13a3f0a675 : SUCCESS in 12m 26s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/b9940e81414a4aa3aa35b0ed27860857 : FAILURE in 1h 16m 25s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e92d0a39a66340d6bcf385568390f9d3 : FAILURE in 1h 28m 22s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/11c7d819052f42328db485805e905d5e : FAILURE in 1h 26m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc6c748dbca74245911c5669deaa02ba : FAILURE in 1h 22m 51s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/485c45b921e041a4954ef2ad8c87f554 : FAILURE in 1h 33m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebbc23e274434827877b1ee184194b47 : FAILURE in 1h 33m 45s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dee48cb397de439eb86d6467d1d90d19 : FAILURE in 2h 39m 13s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36881bb0385f466481776e9e4a638d5d : FAILURE in 2h 18m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0cf3feabda34da6ad30c3cb6649c286 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d1fdd471358f8f876c3937b15bed9895876f4bb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"date":"2023-09-13 13:09:15.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9a50b19d91140fdae4dbed2e632e3b61ff30dd46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"date":"2023-09-13 13:41:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5d6aaca152e01e31bb7d07d893882a0c950c22e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"date":"2023-09-13 13:46:40.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"376a2940a0ebe205b979e09bf2c80eb69c145508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 16:59:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c441f003ea94e05be503f6b5f1187bd\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d66fde43d69a4dcfa88ac134632d7886 : SUCCESS in 18m 08s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4b00cfdda7b445d391615958c7834a1b : SUCCESS in 1h 50m 05s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/67720ab97c61402ea17ce3d4badfbb63 : SUCCESS in 1h 48m 37s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/519b3a105c6643c286917e021d2670d5 : SUCCESS in 1h 58m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e8a9ae6d8a344f38cfd3251f3ae4569 : SUCCESS in 1h 40m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56c5b483ee5b4dbab20de0442829eb3b : SUCCESS in 2h 01m 47s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99f45dd9f7f54743a9b37d82c289253b : SUCCESS in 1h 48m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e91c355f2cfb40b0bdf06194729e5aa9 : TIMED_OUT in 3h 05m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/31f5a62ed8a4404981e1a55ef02dee6d : SUCCESS in 2h 48m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a56c46d02fc46b4a2f4931b345e0d20 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":5},{"id":"93601e2790e628b381c37e827792507b81efcb56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"date":"2023-09-17 07:51:46.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent tree, and commit message as Patch Set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"7778d0e436954edcb58c05dc2520b990844b0da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-17 11:02:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/95afa2f1a84a49b3b09bd950917f1cac\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14a08a47a59947a9970d20c0fadcaf1b : SUCCESS in 11m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2b149d9a7a534b769309119227e6fbac : SUCCESS in 1h 09m 40s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e34aaacc0949448f8661cfcd4911f126 : SUCCESS in 1h 21m 55s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d56793a92b604fcc97251f33b164bbb6 : SUCCESS in 1h 53m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21631726902046ce80a3497b05523928 : SUCCESS in 1h 40m 52s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f459a1f79a464dac96ca9d82fb1a0aa3 : SUCCESS in 1h 39m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10ce60e3b4b548ee934f0431bb936eb4 : SUCCESS in 1h 19m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/974e586fe093483cb6abef5648ce0f68 : TIMED_OUT in 3h 04m 09s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ac486f872e36416eb388ef420c2d9676 : SUCCESS in 1h 55m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18667c9c5a5141d992d16bd74fc09014 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":6},{"id":"2aea23f882e698a249b21058f8a57662fd0af686","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-28 09:08:19.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/c/openstack/openstack-ansible-os_gnocchi/+/895542","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"80aba73d39e8a9d4e752f374d35411481b390104","revisions":{"d66e636695298ef875885e06d96893ce4c1091f6":{"kind":"REWORK","_number":1,"created":"2023-09-12 14:58:40.000000000","uploader":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"ref":"refs/changes/85/894785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/85/894785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/1"}}},"commit":{"parents":[{"commit":"069d0498f61b1ade74108aed5ba700c7f85dcfe6","subject":"Merge \"Fix linters and metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/069d0498f61b1ade74108aed5ba700c7f85dcfe6"}]}],"author":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"committer":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"subject":"use policy.yaml instead of policy.json due to latest change on gnocchi","message":"use policy.yaml instead of policy.json due to latest change on gnocchi\n\nChange-Id: If05936d25db3329b4204172f3b800ddf3417b9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d66e636695298ef875885e06d96893ce4c1091f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d66e636695298ef875885e06d96893ce4c1091f6"}]},"branch":"refs/heads/master"},"f022b2345ed6cb77f90d0894d4dc5b6ee92f86cb":{"kind":"REWORK","_number":2,"created":"2023-09-13 09:33:42.000000000","uploader":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"ref":"refs/changes/85/894785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/85/894785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/2"}}},"commit":{"parents":[{"commit":"069d0498f61b1ade74108aed5ba700c7f85dcfe6","subject":"Merge \"Fix linters and metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/069d0498f61b1ade74108aed5ba700c7f85dcfe6"}]}],"author":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"committer":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-13 09:30:24.000000000","tz":210},"subject":"use policy.yaml instead of policy.json due to latest change on gnocchi","message":"use policy.yaml instead of policy.json due to latest change on gnocchi\n\nChange-Id: If05936d25db3329b4204172f3b800ddf3417b9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f022b2345ed6cb77f90d0894d4dc5b6ee92f86cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f022b2345ed6cb77f90d0894d4dc5b6ee92f86cb"}]},"branch":"refs/heads/master"},"f1f47cad9e746ee3d9baaec5128dcf333ccea69b":{"kind":"NO_CHANGE","_number":3,"created":"2023-09-13 13:09:15.000000000","uploader":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"ref":"refs/changes/85/894785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/85/894785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/3"}}},"commit":{"parents":[{"commit":"069d0498f61b1ade74108aed5ba700c7f85dcfe6","subject":"Merge \"Fix linters and metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/069d0498f61b1ade74108aed5ba700c7f85dcfe6"}]}],"author":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"committer":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-13 13:08:58.000000000","tz":210},"subject":"use policy.yaml instead of policy.json due to latest change on gnocchi","message":"use policy.yaml instead of policy.json due to latest change on gnocchi\n\nChange-Id: If05936d25db3329b4204172f3b800ddf3417b9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f1f47cad9e746ee3d9baaec5128dcf333ccea69b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/f1f47cad9e746ee3d9baaec5128dcf333ccea69b"}]},"branch":"refs/heads/master"},"020da10760bc5b8e880e7a22f72df41ea31677e5":{"kind":"REWORK","_number":4,"created":"2023-09-13 13:41:44.000000000","uploader":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"ref":"refs/changes/85/894785/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/85/894785/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/4"}}},"commit":{"parents":[{"commit":"069d0498f61b1ade74108aed5ba700c7f85dcfe6","subject":"Merge \"Fix linters and metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/069d0498f61b1ade74108aed5ba700c7f85dcfe6"}]}],"author":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"committer":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-13 13:41:30.000000000","tz":210},"subject":"use policy.yaml instead of policy.json due to latest change on gnocchi","message":"use policy.yaml instead of policy.json due to latest change on gnocchi\n\nChange-Id: If05936d25db3329b4204172f3b800ddf3417b9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/020da10760bc5b8e880e7a22f72df41ea31677e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/020da10760bc5b8e880e7a22f72df41ea31677e5"}]},"branch":"refs/heads/master"},"b5db167d61dc778cfc621b43f1fd9de6f5cd7a11":{"kind":"NO_CHANGE","_number":5,"created":"2023-09-13 13:46:40.000000000","uploader":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"ref":"refs/changes/85/894785/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/85/894785/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/5"}}},"commit":{"parents":[{"commit":"069d0498f61b1ade74108aed5ba700c7f85dcfe6","subject":"Merge \"Fix linters and metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/069d0498f61b1ade74108aed5ba700c7f85dcfe6"}]}],"author":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"committer":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-13 13:46:28.000000000","tz":210},"subject":"use policy.yaml instead of policy.json due to latest change on gnocchi","message":"use policy.yaml instead of policy.json due to latest change on gnocchi\n\nChange-Id: If05936d25db3329b4204172f3b800ddf3417b9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b5db167d61dc778cfc621b43f1fd9de6f5cd7a11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/b5db167d61dc778cfc621b43f1fd9de6f5cd7a11"}]},"branch":"refs/heads/master"},"80aba73d39e8a9d4e752f374d35411481b390104":{"kind":"NO_CHANGE","_number":6,"created":"2023-09-17 07:51:46.000000000","uploader":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"ref":"refs/changes/85/894785/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/85/894785/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/85/894785/6"}}},"commit":{"parents":[{"commit":"069d0498f61b1ade74108aed5ba700c7f85dcfe6","subject":"Merge \"Fix linters and metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/069d0498f61b1ade74108aed5ba700c7f85dcfe6"}]}],"author":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"committer":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-17 07:51:24.000000000","tz":210},"subject":"use policy.yaml instead of policy.json due to latest change on gnocchi","message":"use policy.yaml instead of policy.json due to latest change on gnocchi\n\nChange-Id: If05936d25db3329b4204172f3b800ddf3417b9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/80aba73d39e8a9d4e752f374d35411481b390104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/80aba73d39e8a9d4e752f374d35411481b390104"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
