)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~895542","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Ie11f31e6e4a0ee64e6e191820ef9a00e72826a96","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"bugfix_change_policy_format","attention_set":{},"removed_from_attention_set":{"36356":{"account":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"last_update":"2023-11-30 00:40:34.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-11-30 00:40:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie11f31e6e4a0ee64e6e191820ef9a00e72826a96","subject":"use policy.yaml instead of policy.json due to latest change on gnocchi check the config","status":"MERGED","created":"2023-09-17 12:24:54.000000000","updated":"2023-11-30 00:41:28.000000000","submitted":"2023-11-30 00:40:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895542-bugfix_change_policy_format","meta_rev_id":"790f1acc4d4796fd8bdf2f85321564174343ad2d","_number":895542,"virtual_id_number":895542,"owner":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-30 00:40:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2023-11-28 09:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-29 23:01:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-29 23:01:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-17 15:35:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 09:08:36.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-11-29 23:01:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"184754a8673743dfd72b5695763bd8dfc5b356ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"date":"2023-09-17 12:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a24c7bb047240b1c60163bb922224d77e2bcf5f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-17 15:35:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ac8d96de34747afb311e1f36c092a45\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5b488751aed4b17b14e2963b7b1e135 : SUCCESS in 9m 50s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bf9a1d108b5b4eafb2e8e088bab2ee94 : SUCCESS in 1h 07m 55s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d63b0011738d4d9881346240cf2be346 : SUCCESS in 1h 23m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4c4bac08823f4a528d6d95bf1df95ff1 : SUCCESS in 1h 54m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/223aa6b3c2c54c25bff1088279e4848e : SUCCESS in 1h 14m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/529eb2bf11724195ab055b482acbd061 : SUCCESS in 2h 06m 56s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a17b9e758f34385b0bd56d8a36aa3db : SUCCESS in 1h 19m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d45ea955a4fb4710966e2299c678a25b : TIMED_OUT in 3h 04m 23s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37595721a2264ce4beefe61787c7d750 : SUCCESS in 2h 00m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8785c977039b4311a9c65fe4da7e017e : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d3ba41bb5cfdf57f857ccefe090f91f28a58c0fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-28 09:07:25.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"acbff1543647e05e580720ae2ff1855ac0c9936c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-28 09:08:33.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d94857b2ca36e642d50d8a7ed387ad39d5f4a52a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-11-28 09:08:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"21e8d2e37b70f04349764b33d4e36337c2d23796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 12:16:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e2480e4e11b42f2844d0b2e7a097231\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a2d5245a9c14ea9917f3c49676a0344 : SUCCESS in 12m 58s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d14b8e42c6504b8b9c0af96c14b698ca : SUCCESS in 1h 18m 10s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5e01b43c5d044afcb57135638eb3c09d : SUCCESS in 1h 20m 17s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2434b965d7043db97d4c07b9332ea45 : SUCCESS in 1h 31m 25s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c76dd7860cbd4d8f995fffe87a73626a : SUCCESS in 1h 23m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/867a62ced27d42459222c3b7ad97ac49 : SUCCESS in 1h 30m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ae9cff6345ce4038b5d4067cf7cfd235 : SUCCESS in 1h 44m 17s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ac58a9d190e4a3b8bbb267884bd96ad : SUCCESS in 1h 35m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/59ded957e5a74fa9aa40d02981bc0557 : SUCCESS in 2h 58m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44b6ca9a88ab45a2a19019c1a84ee59c : SUCCESS in 2h 46m 04s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e343cdab96184925aaf833355baff535 : SUCCESS in 2h 50m 51s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75696998b23d42cf9f90a86df1f1b139 : SUCCESS in 2h 35m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e64a2aad54c4a8e8b2bd86bb7f91d64 : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":3},{"id":"82b33b9743a39d4431fb42fe5121a21eaaf79ef3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-11-29 23:01:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"48791316f443514fb9afb1b6009588f99643a667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 23:04:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f99941f44b53865db00b9492e2e787caf23adb73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:40:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086c599958a04fe993605ebf76c10c1c\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f80a9430e3394fddaee4d94de6a281cb : SUCCESS in 10m 49s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4e776a0209df456c8fafab210bb012e7 : SUCCESS in 1h 29m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/2a0658ec1dd44191997e32d8fba2d24c : SUCCESS in 1h 29m 01s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/5254771177804553b557f4bcfe34b15b : SUCCESS in 1h 30m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48a342f5845e4d8c9855032aead4be34 : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8927e3ccd51a4152b5db25baadea07c4 : SUCCESS in 1h 30m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d689496c670b4fe5a1e35f3b4005d9f7 : SUCCESS in 1h 18m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00660e26e8a24d3d9661db5b4cd1f41e : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"4ce0bc235dc35390fb7d3570b68c974b98e4a561","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:40:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"790f1acc4d4796fd8bdf2f85321564174343ad2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:41:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d45fe362222b4f468ab6b25427ee5e02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6450c690ac854fb285399102a75cc210 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d44b25ebb804af75dae8e02b36b79493a7e84607","revisions":{"d4bb9e079d315bd560e6f18868b935e646c64e07":{"kind":"REWORK","_number":1,"created":"2023-09-17 12:24:54.000000000","uploader":{"_account_id":36356,"name":"Somaye Ahmadi","display_name":"somayeh hajiahmadi","email":"somaye.hmd@gmail.com","username":"somayeh_hajiahmadi"},"ref":"refs/changes/42/895542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/42/895542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/1"}}},"commit":{"parents":[{"commit":"069d0498f61b1ade74108aed5ba700c7f85dcfe6","subject":"Merge \"Fix linters and metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/069d0498f61b1ade74108aed5ba700c7f85dcfe6"}]}],"author":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"committer":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-17 12:24:24.000000000","tz":210},"subject":"use policy.yaml instead of policy.json due to latest change on gnocchi check the config Change-Id: If05936d25db3329b4204172f3b800ddf3417b9b7","message":"use policy.yaml instead of policy.json due to latest change on gnocchi\ncheck the config\nChange-Id: If05936d25db3329b4204172f3b800ddf3417b9b7\n\nChange-Id: Ie11f31e6e4a0ee64e6e191820ef9a00e72826a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d4bb9e079d315bd560e6f18868b935e646c64e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d4bb9e079d315bd560e6f18868b935e646c64e07"}]},"branch":"refs/heads/master"},"661119895af2b3d4331bb76696d67f0f0c9ea65d":{"kind":"REWORK","_number":2,"created":"2023-11-28 09:07:25.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/895542/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/42/895542/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/2"}}},"commit":{"parents":[{"commit":"069d0498f61b1ade74108aed5ba700c7f85dcfe6","subject":"Merge \"Fix linters and metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/069d0498f61b1ade74108aed5ba700c7f85dcfe6"}]}],"author":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-11-28 09:07:25.000000000","tz":0},"subject":"use policy.yaml instead of policy.json due to latest change on gnocchi check the config Change-Id: If05936d25db3329b4204172f3b800ddf3417b9b7","message":"use policy.yaml instead of policy.json due to latest change on gnocchi\ncheck the config\nChange-Id: If05936d25db3329b4204172f3b800ddf3417b9b7\n\nChange-Id: Ie11f31e6e4a0ee64e6e191820ef9a00e72826a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/661119895af2b3d4331bb76696d67f0f0c9ea65d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/661119895af2b3d4331bb76696d67f0f0c9ea65d"}]},"branch":"refs/heads/master"},"d44b25ebb804af75dae8e02b36b79493a7e84607":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-11-28 09:08:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/895542/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/42/895542/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/42/895542/3"}}},"commit":{"parents":[{"commit":"069d0498f61b1ade74108aed5ba700c7f85dcfe6","subject":"Merge \"Fix linters and metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/069d0498f61b1ade74108aed5ba700c7f85dcfe6"}]}],"author":{"name":"somayeh_hajiahmadi","email":"somaye.hmd@gmail.com","date":"2023-09-12 14:55:33.000000000","tz":210},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-11-28 09:08:32.000000000","tz":0},"subject":"use policy.yaml instead of policy.json due to latest change on gnocchi check the config","message":"use policy.yaml instead of policy.json due to latest change on gnocchi\ncheck the config\n\nChange-Id: Ie11f31e6e4a0ee64e6e191820ef9a00e72826a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d44b25ebb804af75dae8e02b36b79493a7e84607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d44b25ebb804af75dae8e02b36b79493a7e84607"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
