)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~919680","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~Iee1068d98f6b4f3c5415b8052a9876e41c1b0747","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa/rmq-migrate","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2024-05-21 13:22:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iee1068d98f6b4f3c5415b8052a9876e41c1b0747","subject":"Add tag to enable targeting of post-install config elements only","status":"MERGED","created":"2024-05-15 08:16:56.000000000","updated":"2024-05-21 13:23:19.000000000","submitted":"2024-05-21 13:22:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919680-osa/rmq-migrate","meta_rev_id":"71f517350354a043716d2f443598f15c0a1eb25a","_number":919680,"virtual_id_number":919680,"owner":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-21 13:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-05-16 08:30:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-21 11:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-21 11:21:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 11:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-16 08:30:23.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2024-05-20 08:00:26.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2024-05-21 11:21:55.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"72950240de2466ddb98c7bcb62e868a2b53a8166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-15 08:16:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38c4a63b49579a4523a8e0fa1f0febddda8f53f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 11:52:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b9e05332e224d2ba0966c9d96fa5160\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb10c559fce7468aa3a8064291cd8d97 : SUCCESS in 21m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9bba1ecf32a642c6aa5a3b297e250a19 : SUCCESS in 1h 24m 47s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/77273186b4bd409eabf9f22d276cb891 : SUCCESS in 1h 25m 58s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e82cded7136145d58aad6528e164989e : SUCCESS in 1h 52m 35s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/74a9c770b30441109a5417fc965d6928 : SUCCESS in 1h 33m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb724c3afcc04a0fa382458ed83a1c4a : SUCCESS in 1h 38m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7926018092da4e24b7e7b9238321ccaf : SUCCESS in 1h 48m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2370b7d7591340fbacb5b67de818e52c : SUCCESS in 1h 37m 55s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28c98c6fdb3c4334a4b12a6e40d50e2f : SUCCESS in 1h 37m 27s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/87a094ee83654c7a84b5a690c225303c : SUCCESS in 2h 51m 53s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0038adc2a23543878c6195bd9225a659 : TIMED_OUT in 3h 04m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5054ff21c85744339122ee3a294331eb : SUCCESS in 2h 17m 37s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2dfe7ca3e8624672bbe8f7060d6cd32a : SUCCESS in 2h 47m 08s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/de2c83c4c747415793a9937b898bda56 : SUCCESS in 2h 39m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c3035ce77274bc2814dd7434cbd2934 : SUCCESS in 2h 29m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac0bef8689b8443d88d32be7b38b7803 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"86536f125955ee1aa1ee6126ddc24badd91829fb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-16 08:30:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c799dda066a46d71a9c67914e3eb5b48a4e4829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 11:44:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbd75ada668040e381281fb0498dd03a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f1f98eb35ae4d50ae4f12f3056355f0 : SUCCESS in 14m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d81f5ca06a2547a2b740e33607bd46aa : SUCCESS in 1h 34m 09s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ef91fe48de02450186d6a440f09146c9 : SUCCESS in 1h 38m 53s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4664097d5d5400b8f1241844adc7866 : SUCCESS in 1h 38m 46s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6a8ab14a79f0428c8e72b52ced445c8b : SUCCESS in 1h 42m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f024c3d16514e64b4fc2ea84d44a0b5 : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7a63c00715e14a98aa039290e62039b6 : SUCCESS in 1h 44m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a6d514de1e843ce8df268434b338b9a : SUCCESS in 1h 32m 18s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d744736db96847c984766d57041d9e52 : SUCCESS in 1h 29m 30s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/81b356c96b694a0fa5627fa718fa3632 : SUCCESS in 2h 35m 58s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2a77f287a8c74f1b90701f1c45362bc2 : FAILURE in 2h 00m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2915ecfc748e4fac821a8e9bb8f8b066 : SUCCESS in 1h 59m 27s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/695841e6123d449bb786e5831d844d6f : SUCCESS in 2h 53m 50s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c375d699895465dad53bbef259051bb : SUCCESS in 2h 19m 09s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc7f72d2c8134dac949602cb9b495e6d : SUCCESS in 2h 20m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd787d79349b4335affa9317d418f004 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bc176f1a337728eec043de66dd5b7f6c6800ddbd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-05-16 13:11:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"882e5774e7a6b5fc6525a0db25fb6768922a8770","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 16:24:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb671376c0ef4ec09ba8926b508decb6\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a3575547f76468cbe5465ea3e6379e3 : SUCCESS in 19m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bfc218e6b20243b3b953ef4625abbab5 : SUCCESS in 1h 35m 26s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6f4dfc3d4e54474bb801961718282aeb : SUCCESS in 1h 40m 48s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2aa984d1acd42c5b9e1956c7128d5c5 : SUCCESS in 1h 49m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2117e3beb24042c9a6d12ced1c986226 : SUCCESS in 1h 53m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da02ffb44e6b4e5383a8c2027add8561 : SUCCESS in 1h 47m 27s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/102e4e638d354587a9754a99ed778fe5 : SUCCESS in 2h 00m 36s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5191180d7c943879783095ec2f22abf : SUCCESS in 1h 31m 24s\n- openstack-ansible-deploy-aio_metal_quorum-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd601b335d0948c6bda7ce87478b0330 : SUCCESS in 1h 37m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bcbe770e86a24d09b59eec63f80f27d2 : SUCCESS in 2h 38m 26s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c66227b8b53b4f60875f2feffddeeafb : SUCCESS in 2h 43m 31s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/692518b1b6b5462b8f30ad296831aae8 : SUCCESS in 2h 10m 01s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8826e1695b9f4b05b103241ebf529b7e : TIMED_OUT in 3h 04m 06s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d6559db85a56418c8ab90237de5fe27e : SUCCESS in 2h 39m 34s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93a7b1d6feb84512b7f4774ec3afd786 : SUCCESS in 1h 55m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1058144a1c0d4a8ba4148ceb8d7cfbd4 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2ce607085d9e39697d612c9975e058cd2c4508a9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-17 07:07:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"226f1a6e90e4bc0a21a25a2749b74a1d5d07448c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 07:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef328841a79e424bb4dc086c6456b247\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/810db1f30880453ab13d27fd220aef59 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"794d725bc58ddd53b6cfc8afd8907c8ddc4989b2","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-17 07:16:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ede12c20c4dc3ed4a0d42c619ef21be8a49ba99","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-20 08:00:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59917d3599e75e60e2c33c8f91820f6a71a90765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 11:12:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d3afb469c934db9b521853f4c4ba18a\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f186c5787ac4431dbe4797e2f68d3a06 : SUCCESS in 12m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/915a0af421e74902b44e3fc6237429fe : SUCCESS in 1h 25m 52s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5fc1239469934af3aeb850beddaaa1fe : SUCCESS in 1h 32m 21s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/63264fce3b1c4d4280918bfe060d15f8 : SUCCESS in 1h 39m 17s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d0e69ce9fb884c348bc93cfa63926bc5 : SUCCESS in 1h 40m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81522520f6984ff7ae45ac479152a080 : SUCCESS in 1h 37m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4b4f0a8333064ac3930f46c950f6a6c4 : SUCCESS in 1h 37m 23s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea07ef9c411445f98e578b0e228dcc57 : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cb97a549e414707b596224e8b884dcf : SUCCESS in 1h 29m 36s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/928ea810d4cd4631b5ad95ae0406f806 : SUCCESS in 2h 29m 51s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/10632b284c7d468495d1a1c1fc2a1be5 : TIMED_OUT in 3h 03m 52s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a60632403a44c4ea4ad270d6949a3d6 : SUCCESS in 2h 12m 14s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/565df8298bcf405eb5e59e3702a4b9d0 : SUCCESS in 3h 02m 20s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8276ecf3a82a47119d47ba7ac07ea49e : SUCCESS in 2h 22m 24s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1fab7cc68f94ed682ddc2478d7a4988 : SUCCESS in 2h 07m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484e57b51cac43c897badd95e6aa45ea : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1b051582e756478258482d6bb0f65eb390b300e1","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-20 13:27:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9a1d5463f80215834846bdd22367d1a4cdfefc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 16:40:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b03227629e8945eeb2804d7903dfa052\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30a8cea2bc754748b66836b302890fb4 : SUCCESS in 12m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2748d9608abf4f7485acf3ec47d53a58 : SUCCESS in 1h 30m 34s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/cce13154247c45808f2719a73cd91097 : SUCCESS in 1h 40m 19s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/847333572fd444d5a22f62b952af86e7 : SUCCESS in 1h 42m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a93e69a744df4b6786fa4a1d2effd14b : SUCCESS in 1h 43m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/361855ea6321468d96652a39d9d51750 : SUCCESS in 1h 29m 39s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c7306da060c54ea1b5e5740d7137da68 : SUCCESS in 1h 48m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5064340883d44a9badac9a1a8923d7fc : SUCCESS in 1h 39m 24s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35f5e7dc08b540c1bd9269aa1f7f80a2 : SUCCESS in 1h 30m 59s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/652d498d7f8342a1b6222528183a5b37 : SUCCESS in 2h 30m 35s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/40c331769fef447585549024927503b3 : TIMED_OUT in 3h 03m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5d376f97efc412e89803361b0ffec98 : SUCCESS in 2h 28m 54s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29f5e9a64861473aa9d6d3520e21c6e4 : SUCCESS in 3h 01m 18s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/776a98157ef34a73bddc2ff38d021225 : SUCCESS in 2h 31m 57s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cab00cb565f346d2b7c66d5d16e3e182 : SUCCESS in 2h 29m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d6d31132a04d9d8dc0b3c4950adaa0 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f913cf454c7877170412ee8da3a6843f98d32230","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-20 16:46:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c85e0ea273bf4def6f74cb711a033231db298e0a","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2024-05-21 08:16:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb91463710bfacff921630fba5be2e833aa77fbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 11:14:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41367ee7e5dd4ce49c9fc641ebfa2127\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca2e805b4b8e4a15b40b824878fea031 : SUCCESS in 12m 41s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9eb321ae05904222b5d4c22c232de327 : SUCCESS in 1h 19m 19s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bcf7d8d1da62402d94f7981ac6308ed1 : SUCCESS in 1h 40m 43s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f90ac78f80b4699a2f89cd75f0135ee : SUCCESS in 1h 34m 48s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fb01d2b7d38742dc9b7b2347831366c1 : SUCCESS in 1h 50m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2dcb927fe644e8dac9064069708b528 : SUCCESS in 1h 26m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a076b9690ff14f469a389792dca3f1d0 : SUCCESS in 1h 51m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b11e1d28137d459dba33a81219110e2a : SUCCESS in 1h 33m 27s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05ca283c00d8425da7737d24265fa74c : SUCCESS in 1h 24m 05s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f2f04dc236bb4d0daf112afee7fa4b29 : SUCCESS in 2h 18m 53s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1b7a33d640bd4fa981f8fabc3f23a38c : SUCCESS in 2h 49m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eef17aab591c4eefac04e999fdd69c3f : SUCCESS in 2h 00m 15s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a38d8cceb284d70a8f31005c312210d : SUCCESS in 2h 39m 14s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cac43b33cfc64dfd9625f6cea615ba52 : SUCCESS in 2h 23m 38s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dec6c5caf93e4622a5491ed6d330e660 : SUCCESS in 2h 31m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/757082474e084fb1a434ee741cf0368f : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7b6afc7f1f3d4462109e161714e2b492ebef5d9a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-05-21 11:21:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e69f2564a1299973d2db6c71c3843a9935ba16f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 11:23:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be8d9916723c2cd99f7079a1189a40e113b7c6e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 13:22:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ef601797aee49d09a5841d12a70faf3\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d88b1c41a3d4eeaaa2a50a54934c7d5 : SUCCESS in 18m 04s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/171dbb7fe9b64538bf8cab36b250eb7d : SUCCESS in 1h 36m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5a99a3920629440a993830885c0c654a : SUCCESS in 1h 42m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/305287dd56984e258de2de496af24b87 : SUCCESS in 1h 48m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9c2f75c8cb04cc1be05560221b3b5c3 : SUCCESS in 1h 15m 05s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e668b8a58b54400fb6a7222c47067ea8 : SUCCESS in 1h 22m 53s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fad435dbc3234fb18120b573dceef632 : SUCCESS in 1h 47m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3643677841e84b1ea8dda64da04f3e5f : SUCCESS in 1h 27m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0c11887bba4151b7620f9264e99ee8 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8ad8b9fd82785b1e53ea3ee20b71c252261e0ec3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 13:22:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"71f517350354a043716d2f443598f15c0a1eb25a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 13:23:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5234ff1311e43e9a4d0bfc55ee6d348\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6cfa18f5c214abfa42865f0b5b6d76f : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e61e48a365044607e8ac57fb1d7a4f12c63f4ea7","revisions":{"e61e48a365044607e8ac57fb1d7a4f12c63f4ea7":{"kind":"REWORK","_number":1,"created":"2024-05-15 08:16:56.000000000","uploader":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"ref":"refs/changes/80/919680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/80/919680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/80/919680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/80/919680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/80/919680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/80/919680/1"}}},"commit":{"parents":[{"commit":"7e2a4648486804fd95babbd4fd14696f4f5a5cf1","subject":"Ensure Gnocchi is connected to MySQL coordination with TLS","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/7e2a4648486804fd95babbd4fd14696f4f5a5cf1"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:09:18.000000000","tz":60},"committer":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2024-05-15 08:16:54.000000000","tz":60},"subject":"Add tag to enable targeting of post-install config elements only","message":"Add tag to enable targeting of post-install config elements only\n\n\u003cservice\u003e-config tags are quite broad and have a long execution\ntime. Where you only need to modify a service\u0027s \u0027.conf\u0027 file and\nsimilar it is useful to have a quicker method to do so.\n\nChange-Id: Iee1068d98f6b4f3c5415b8052a9876e41c1b0747\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e61e48a365044607e8ac57fb1d7a4f12c63f4ea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/e61e48a365044607e8ac57fb1d7a4f12c63f4ea7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
