)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~921284","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~stable%2F2024.1~Ie7923f3b1fa3fcf3f463b476f908be09533e103a","project":"openstack/openstack-ansible-os_gnocchi","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2024-06-06 16:32:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie7923f3b1fa3fcf3f463b476f908be09533e103a","subject":"Drop default policy file location","status":"MERGED","created":"2024-06-04 13:01:37.000000000","updated":"2024-06-06 16:32:59.000000000","submitted":"2024-06-06 16:32:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921284","cherry_pick_of_change":913244,"cherry_pick_of_patch_set":6,"meta_rev_id":"c0f29965ef67b01feee2682a37812eda949fed46","_number":921284,"virtual_id_number":921284,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-06 16:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2024-06-05 07:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2024-06-06 14:27:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-06 14:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-06 14:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-04 13:10:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-06 14:27:05.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"},{"updated":"2024-06-06 14:39:34.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"11b36ee366a893613172c7041a4e65bbb9cdf06a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-04 13:01:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0a4f23da6672e69552b0c8a0e4992a3b742404d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 13:10:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1340ff14ebc44e639f462871662c7f85\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950dc54a0176403dad64aea9558ca5ba : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d33dde4c964e4470b5951c0b11b6a144 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"45a251e1239a6b56ec162f16380d8e9667517ad2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-05 07:35:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2b14554be138ae3f774eea9a20cea96e17542eff","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-05 09:08:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d86a1ad7771129d4ae2dbbf1fc3534dcd61d936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 12:16:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba166f96ae9e4a33b9fb9ded2c5c7ce8\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/278dba4e60d74ee7926bdf92b3043f5e : SUCCESS in 12m 43s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/667106c7fafa4a74bf8621b67722b645 : SUCCESS in 1h 40m 12s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/bf6024011d1145179050f8ce8da9348a : SUCCESS in 1h 34m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e09d1901974c4fb0a7f2a2f5b728a327 : SUCCESS in 1h 20m 29s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a9e8d2a0a4e8426392b01ad392ce5b56 : SUCCESS in 1h 51m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a18bc8154cc44834b206ce6a7db05973 : SUCCESS in 1h 43m 02s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/298a6cc208bc40cd8feb9637358e4ae0 : SUCCESS in 1h 57m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbe1586065cf446e82178fa64981e9af : SUCCESS in 1h 45m 57s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/913a5352eea24319b3893c35509daef3 : SUCCESS in 1h 39m 22s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4514d135b1c54d63a4f389b9ebea384d : FAILURE in 24m 11s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/18d478beec4345e3997d1c6d685d1d51 : TIMED_OUT in 3h 04m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a1de6cce0b13470cb4bee2413847ee1f : FAILURE in 15m 54s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/072690877c4a4405aa709251a37436c0 : SUCCESS in 2h 55m 36s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c396523a2f0541d4a93741847e92e0b8 : FAILURE in 16m 33s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a9ecf31cae14459ae69ce37a968b746 : FAILURE in 17m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a0c4fc56b9b41a4bdb2b7b239e241fb : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5eac2a0f3ebd4741bd832017d7047382 : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ce2671360f29a74632920f11b7bcb8ab8d9f40b7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-06 07:10:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e972e8ed23883f84c6bb45e0290e936175f829cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 10:24:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/457f04d0030b430a8d2b5d0bbb47a2a2\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27b11067f7a843888f41afee07343064 : SUCCESS in 12m 11s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/52bcac1f7d254c9486dfced1fea59d0c : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ead0014ffe874bc68ef98a9bc96cedb8 : SUCCESS in 1h 25m 37s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/570a9ce629c94e5b8049b6415169d362 : SUCCESS in 1h 45m 20s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6c663c920ea84e0fb7725c281ba1cd55 : SUCCESS in 1h 31m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4e045cc50622450da4811a3e4f606cb5 : SUCCESS in 1h 49m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d43327d11cf4eb7bda4798764af4087 : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/063e52a1441a495ea88c8faafe6c86dd : SUCCESS in 1h 42m 24s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d56ad97043274d62a13d836ae8d3b3bc : SUCCESS in 1h 46m 24s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3ceabaed8ef54193940ece05877294dc : SUCCESS in 2h 44m 23s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1d4775f535774e2f8a1699a392d58a53 : TIMED_OUT in 3h 03m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/391ab32c4ba846568782d9688db33334 : SUCCESS in 2h 10m 24s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d80b0801cb274edab52a3d4df2a6f9b6 : TIMED_OUT in 3h 03m 34s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1255b5e7aa664c58bb0eac8e723c1dc7 : SUCCESS in 2h 48m 37s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa13ef66ad224314a6818b922f9c8a00 : SUCCESS in 2h 37m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb1967b80f942b681909d6270e83dba : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85b88ad78a0648b292c3b293ba6ad1aa : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"721611f722012da100f86dc466d5e5409b5852cd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-06-06 11:13:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c715a291d8c2ffe2f649fb46cc89b2a8f695a18f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 14:19:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9f113fc0d6f4a589bc3e9d8b5d6334b\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e277a22a1b4b41a5aec69037876b3466 : SUCCESS in 11m 20s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d85472ea301c49c1856470818b189af5 : SUCCESS in 1h 18m 46s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/16a9204acbef4306ba571484dcc0c5b1 : SUCCESS in 1h 32m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3e32f95e1fb4e98a102487f78397a8d : SUCCESS in 1h 58m 36s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8214c8d14c534914ac54117023d7c0e2 : SUCCESS in 1h 31m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f96323d00e81432d8e0e15c10dc0df46 : SUCCESS in 1h 19m 47s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fcf92b14f8f8499887e28fbdbc4b52fd : SUCCESS in 1h 41m 41s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a6fa7326c7344c7b04453074fbbcd7c : SUCCESS in 1h 27m 22s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a522ccadee984d4db9a9721955d91380 : SUCCESS in 1h 23m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e4a84310014f4a7cb3fdf91d2cd0c9b4 : SUCCESS in 2h 36m 42s\n- openstack-ansible-upgrade_2023.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/bb5969b7b3b54961bdcd8385908d3835 : SUCCESS in 3h 02m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40c817081e6d4e7f8b40d9bfc3203274 : SUCCESS in 2h 20m 20s\n- openstack-ansible-upgrade_2023.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6c3252d466e4d118270f7884b8165a3 : SUCCESS in 2h 34m 49s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e855922117bc42ba8176c05542696b7c : SUCCESS in 2h 50m 30s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2528747044f14e1ab4ee4cb47f8a681c : SUCCESS in 1h 56m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf48a769bce349df95e456866c24a575 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea158845a5f441bca1f7b3aafd9e0757 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ac6caa4239cb1b2d4eb2badc522b7bf3259ed5e5","author":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"date":"2024-06-06 14:27:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3909b796fbc75ea9e8953f76322787be4de6725e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2024-06-06 14:39:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d47e8f3b41342b0a25cb0e1d8617a3ac88eae28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 14:41:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4043751fb5f3f7348450c0b481084f592791868","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 16:32:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c21393718fd54f41baae17f867c752c9\n\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de5023dcc0f14e63a365174aa75865e9 : SUCCESS in 10m 07s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/9a0fb41551bd4768812128f14282c21f : SUCCESS in 1h 21m 42s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d20167943c48468dace639a19a706e43 : SUCCESS in 1h 37m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/aa09cecdaa95498a9184910ecfd7cf56 : SUCCESS in 1h 38m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71df500f51b24488acfee2488cd91764 : SUCCESS in 1h 35m 01s\n- openstack-ansible-deploy-aio_metal_haqueues-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/132401b1a2184b799e2a40db7e2435a1 : SUCCESS in 1h 33m 00s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7c55cb189cb345ffb426a36fae5fcb45 : SUCCESS in 1h 45m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14660b09ae9d4372a19800e79a63f3e8 : SUCCESS in 1h 34m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbba0727b23749888ff48faa4789fd16 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3ca09e654bd4891a1634a5f064d4df9 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bc988d381f1bd3add7380e744041d219c92b36be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 16:32:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c0f29965ef67b01feee2682a37812eda949fed46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 16:32:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d5392bf2eb9401f81c03cb32a5e3fb3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7797ec246b294c38b091a510135bd745 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be6973bb2bca42b05c15327e0105bd060b06335e","revisions":{"be6973bb2bca42b05c15327e0105bd060b06335e":{"kind":"REWORK","_number":1,"created":"2024-06-04 13:01:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/84/921284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/84/921284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/84/921284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/84/921284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/84/921284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/84/921284/1"}}},"commit":{"parents":[{"commit":"7cce9921b400a65ddb7e8fc07d1753e2131246dd","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/7cce9921b400a65ddb7e8fc07d1753e2131246dd"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-03-14 17:36:34.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-06-04 13:01:37.000000000","tz":0},"subject":"Drop default policy file location","message":"Drop default policy file location\n\nWith master version of Gnocchi[1], default policy file is no longer provided\nsince all policies are currently stored in code.\n\nMoreover, it\u0027s long due to switch policy formats to YAML instead of\njson.\n\n[1] https://github.com/gnocchixyz/gnocchi/commit/6ad4bb8e62f2ab6f5e80e94d0e055a748f914561\n\nChange-Id: Ie7923f3b1fa3fcf3f463b476f908be09533e103a\n(cherry picked from commit 8c503a40c19d548d47b76331150aa5cba99f0112)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/be6973bb2bca42b05c15327e0105bd060b06335e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/be6973bb2bca42b05c15327e0105bd060b06335e"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
