)]}'
{"id":"openstack%2Fopenstack-ansible-os_gnocchi~981602","triplet_id":"openstack%2Fopenstack-ansible-os_gnocchi~master~I62b8bdd569ba54e8d18450286234617ff1ce254e","project":"openstack/openstack-ansible-os_gnocchi","branch":"master","topic":"osa_hashi_vault","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2026-04-21 01:57:55.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-21 01:57:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I62b8bdd569ba54e8d18450286234617ff1ce254e","subject":"Add hashi_vault pki backend support","status":"MERGED","created":"2026-03-20 20:26:02.000000000","updated":"2026-04-21 01:59:02.000000000","submitted":"2026-04-21 01:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981602-osa_hashi_vault","meta_rev_id":"1c300bb79018782ebfb6b7b4b505d3092ce073d6","_number":981602,"virtual_id_number":981602,"owner":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-21 01:57:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-04-20 12:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-04-20 09:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-04-20 12:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-20 23:13:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-18 08:11:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-04-20 09:20:16.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2026-04-20 09:23:04.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-20 12:42:49.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"4898a5b3a1fbeebe02f272c971ecb8c12a9fd7c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-20 20:26:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa269e5371cb744f66d920dd164e2cb1f8f64e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-20 23:13:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4467c3528077413ba759d7820023d335\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/81d778db094b44bb8d2cf44593510f4b : FAILURE in 12m 49s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c7b15596a2774a9da5f29f638892ec47 : FAILURE in 1h 27m 17s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/9820b11a9c424d92ab91de6c26ee60da : FAILURE in 1h 18m 59s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/228e13204f3c4745baee3ff9a6e87f8f : FAILURE in 1h 27m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1df24e940d484f8eb6bfeefe6ccfa933 : FAILURE in 1h 30m 06s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8f60bd43e99a4bc79b4a4093c7b56638 : FAILURE in 1h 26m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6fd7869bc5c446f696614fdb763dc540 : FAILURE in 1h 18m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a1f5edcce2d4239ad3ef4360489222e : FAILURE in 1h 37m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d6ddd8faa9d8463d9dca701f0f06ce24 : FAILURE in 1h 54m 29s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f5d735f682144302af26e5a2fb52abc2 : FAILURE in 1h 16m 26s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7679cdf195c1423b93cd67276df9d98a : FAILURE in 2h 25m 19s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/175d4bb49dfc4e50800b3bb763a65279 : FAILURE in 2h 17m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c7153f5d19774fda90478c99c2347a05 : FAILURE in 2h 28m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc884e0ed36d460a87af2dc27180ffcc : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":1},{"id":"eb151a7c973471bcf5b99e60670d0ac740ef8e00","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-03-31 11:05:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7344ca8047b28676895bbeb18c4e7304eeffdfe4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 14:05:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/51b58bc0c8ab4bddb31065bae240f33c\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b0af12f2c8f24922af5acdbdbd05b9f1 : FAILURE in 12m 51s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/42af9bff56124f2bb28c3af0c2daacc8 : FAILURE in 1h 21m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/6611e9607e054aa1b16414bf82fca263 : FAILURE in 1h 22m 12s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/c5f38d6bb95248748af735e0338f0a0e : FAILURE in 1h 28m 48s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fc76c22a583b43fcaa8a55236199efcb : FAILURE in 1h 33m 28s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d14ff97f7a954a00b397e5c44c2d9568 : FAILURE in 1h 25m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/07c1db4aa9c34bfebb9540ecaf23c1e2 : FAILURE in 1h 27m 49s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2d96ae539af845119e880e95b41f8cb0 : FAILURE in 1h 32m 10s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8874265bba9a42ba833fc628003247e7 : FAILURE in 1h 57m 26s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0418c03f57264792babe20096b73b202 : FAILURE in 1h 38m 37s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66980a023c51429fa5672bc16c27b72f : FAILURE in 2h 24m 56s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33d8cbbf8eeb4369a9b66748ac29050f : FAILURE in 2h 41m 47s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f31692f8bfa649bb9a352f5d31ddc0df : FAILURE in 2h 31m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8704443d7efc4e5abff2fc8291bbc3c1 : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e808344830715d9ee6567981410419f15fac3ed8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-10 07:40:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_32666\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}],"_revision_number":2},{"id":"d11b3a074958b9e92fec70e152611f20dd4d70f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 10:20:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50872ea7411f4bc1ae6dd347128c1795\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4494c334b744a859e9b6a2acc397d94 : FAILURE in 8m 58s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/8a2af0d9adda4e6f8d24b15b593d8c43 : FAILURE in 1h 26m 04s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4ac620872ca04b55b3977dd0e4266391 : FAILURE in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d40c9fd05b674e138d1fa8af67cc2778 : FAILURE in 1h 15m 07s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/a2ce37c97a5e4cec8f3814e05da548e0 : FAILURE in 1h 36m 25s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9c2802f2389b487d81d183616ab729e6 : FAILURE in 1h 39m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0065dddd2221413fbdba9916b1de8c4a : FAILURE in 1h 23m 26s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37a429a76f6345d38d2f9950487266fa : FAILURE in 1h 33m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/7ee74fd4c9594f9abae9d13474f0e7aa : FAILURE in 1h 10m 52s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/cbfd265233b644fcb8ac51735b110d43 : FAILURE in 2h 31m 23s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f5de94b3e99d484abb5fc06a30ce2e88 : FAILURE in 2h 21m 09s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af4da4a766d8491ca11f22263ecea4e7 : FAILURE in 2h 21m 58s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b9f867fcc1d8402bb23c6295461f9d4b : FAILURE in 2h 22m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75bb0b12b69848aaa1ebf99df8d56a35 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2ef9a09e4c728edceff574ce654fda4e6887b2db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-15 08:59:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"2a6c67839da8a7acbaba0f135441c5edee7122e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 12:05:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d0620a747f04170b23ede6276ced5b9\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c75678acd2bb451a8a678a97ac26fed2 : FAILURE in 11m 01s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5e8fb9cb10be4a629b9b855490afe3ab : SUCCESS in 1h 23m 48s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/4c65423c850e4e49bb3b85e00f66a26e : SUCCESS in 1h 39m 55s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/cc00a7ba9e50406f94ce5cc21e8051a0 : SUCCESS in 1h 48m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/00cd8673f498461b9b2a3679b8e4a792 : SUCCESS in 1h 32m 32s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6b7f5f83ec364edc9b41375a5af6a939 : SUCCESS in 1h 52m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a4909d8fbc0456c828e08b2d0c94500 : SUCCESS in 1h 48m 57s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d7b1529dfcad411fa974887fb8e27f75 : SUCCESS in 1h 51m 50s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/67790c4ec3724ce492a6ce417bed7a12 : SUCCESS in 1h 25m 55s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6decc151cfd5410a97718609dc7931b0 : SUCCESS in 2h 48m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/64d861ba4eed4d64bdc5a3e07573190a : SUCCESS in 2h 37m 27s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/10af433a8f574401914c260de3a4bbb5 : SUCCESS in 2h 47m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c385454f0e0c4d4e8884b7fbe9f4e101 : SUCCESS in 2h 56m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78b2828c41464c3d905401ada7b14274 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":3},{"id":"1fd451833956d733efc02617e32190c16a207c17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-15 12:44:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c1344d25ce0c79cfd3ffbc62e316074920f5f6bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:42:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/818fcdcb60ce4e77bb0360741f3c4d5f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1717a15944084339a920ef21ece66c74 : SUCCESS in 12m 58s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/70b6e2b4b05945359473d37c4321cd30 : SUCCESS in 1h 03m 33s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/ad44e2f12c9c4d12aa1a15ab7cae0dd6 : SUCCESS in 1h 41m 00s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/b8b48af41f8845caa9cb8dec479e700b : SUCCESS in 1h 36m 55s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/bbbdc566901141c4aa5bd4b12a0b6a7f : SUCCESS in 58m 38s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b9bdc3b96dd54f88ab8df1668485ea00 : SUCCESS in 1h 10m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67f22ceb5a844ae1a61c6bffe1fd3d76 : SUCCESS in 1h 00m 46s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4854a1e177eb4daab86ffa0bf1701f84 : SUCCESS in 1h 52m 47s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ac5135b5906a4bbd89fd3e5691b0101a : SUCCESS in 1h 24m 21s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6b268fdabfe74709b46c1b6f623fe1f8 : SUCCESS in 2h 49m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d3372757d0d44a998d1e5925f31fa239 : SUCCESS in 2h 40m 57s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/de02b0bf3c214b2bb62bf4d03f3a67ff : SUCCESS in 2h 43m 42s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d8b358c2acb44c52a28dfcc889fdc2a9 : SUCCESS in 2h 37m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a749858aed14577bbc01647f6e0b911 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":4},{"id":"5bceeba8353b6066094985ae2408c823e31eceaf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2026-04-18 08:11:48.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fff21d48eedb23eaaf854385eb216e42c8e37294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-18 10:47:37.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"623179a1a4cd75ab6c01e67296e8bd06ebef8a6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-18 10:48:20.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"5a49057011c352e0ac113624e0a6cad2d6675027","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-04-18 10:48:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e60e1863871fefaadf961da81d4c9bcf86782760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-18 13:47:17.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/181d733cde8b44219ae7f95db399bf22\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e494a504b03541e79e737685958f9f49 : POST_FAILURE in 12m 06s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/b28a85a5b085454a8bbf5be9cdd86f47 : POST_FAILURE in 1h 02m 26s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/5b35bce1766b4a928c106df10c0e24fb : POST_FAILURE in 1h 36m 54s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/3c1be191b58945558c2d1f02e028a09a : POST_FAILURE in 1h 40m 32s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/080a20c9f38e40abb8b3890d32bb68b4 : POST_FAILURE in 1h 02m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0409ccaef4ba47f19795c359ba2dc6f3 : POST_FAILURE in 59m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/82d2b025a2c64093a52ab12ebc892a6a : POST_FAILURE in 1h 40m 55s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e65da8f5ab0b4210bd674f26a9598ff9 : POST_FAILURE in 1h 42m 45s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3bc19cf9ce5e47fc80cb162529bd4742 : POST_FAILURE in 2h 42m 49s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f03531dfcd7d4808bb000311e29a474f : POST_FAILURE in 2h 36m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5361a941882c4e1186a0f51bc640a10a : POST_FAILURE in 2h 41m 01s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ebb23eb8f6cf49e6a0094d0c451d380d : POST_FAILURE in 2h 42m 36s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9bf51ec319474b1194f8fe89d8eec3bb : POST_FAILURE in 2h 47m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8585b2e305e484381158bf18218db90 : POST_FAILURE in 6m 40s","accounts_in_message":[],"_revision_number":6},{"id":"68658175d23421ff382ab44e54b4841f80aed46d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-20 09:20:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"565f2c0400b4fa7525b822d754ac837f38eaac3b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-20 09:23:04.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"33c800f3a36effd79e8e69619e98d29d1e9343f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-20 11:20:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b9538292f033a85585496d1f4a5a353ea0700e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 12:35:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c1963e81ee144a9b007cd2cb276ab18\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/05da36399bf14af99284a469a6276089 : SUCCESS in 14m 47s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5f38872a52f34de59f5bef920c01ef77 : SUCCESS in 1h 32m 47s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/45e75cfb64394fa7ac4ea09e7789dc6a : SUCCESS in 1h 38m 19s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/040b366d384846bb8dbd9ee295d91d21 : SUCCESS in 1h 44m 02s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/94a4f35ba4734acd9bc51da53ccd9e1c : SUCCESS in 1h 34m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/97a1242436c44e83998062bb1cd35c81 : SUCCESS in 1h 55m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49d6b5b4e501419981b044fd371c0127 : SUCCESS in 1h 35m 02s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3e94b77ede2c4aea84fec039984709ff : SUCCESS in 1h 05m 51s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/d8948711a53d4b3385abcd8e94db03a3 : SUCCESS in 2h 19m 33s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/ef857afe551645a0baeb0b7cda97e5f6 : SUCCESS in 1h 35m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d087f04bbcd46a4ab5c86d7e76670dd : SUCCESS in 2h 57m 21s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/47dee9fa4d2a4dc4baa977fd9630ddb5 : SUCCESS in 1h 25m 17s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7e830ac34f964215ba48b0dfd4e72ece : SUCCESS in 3h 06m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55986a87f734a3baf32ac57d20d8503 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":6},{"id":"0f711b1b9abdc73ea2934a14a4295c00423edfbd","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-04-20 12:42:49.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"85f6d8fc39e3c3f9484ca819baa0917f2fe0a6b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 12:45:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4ca5f2b05d8154378194c97d7a6af16da1b1a620","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 15:26:48.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b5d7251177a45c592a77b54e89ccb84\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/62aa50c3df7047a1b2c0716f089a786b : SUCCESS in 13m 24s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f6534eba64d3415f9f4718c19928580f : SUCCESS in 1h 32m 53s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/06bb3493f7164b7cb3a3226dde17c781 : SUCCESS in 1h 43m 17s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/e592d38879b448cfa5275e1c646e8574 : FAILURE in 1h 26m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/6f80b1061f4048fb8704dfde34c6c66f : SUCCESS in 1h 37m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3ffed3006b484fa199683357cfb27a6a : SUCCESS in 1h 33m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d5bf9f23aa942cead7483fb4fdf29cc : SUCCESS in 1h 43m 40s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33b09ad15b464c0bbf1c836d9c8cecc8 : SUCCESS in 2h 07m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b5ee514524d46659e0115931c0e5ef5 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":6},{"id":"938c1bd30ab4949e17489210fbfbc45240508378","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-20 20:33:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dc70d5b03c5467b70f74010cf7b71c96151a575f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:51:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/144332d4b3634a04b01a43a4a18ac66f\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0ea04c24a3de43f2ba4d746327454552 : SUCCESS in 15m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/842a9bbf8b48409590b18d132df8a558 : SUCCESS in 1h 30m 15s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/77d266b392dc4af480baa9a752fe03ac : SUCCESS in 1h 42m 21s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/265ee867f8ca444a8e36117522e2cce5 : SUCCESS in 1h 38m 49s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/b70f79612579423fae9f6a8c35462dcc : SUCCESS in 1h 00m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/90de83e3f3744d268d2256a685c4b576 : SUCCESS in 1h 02m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5284947b6c347289739c05d6c97dff5 : SUCCESS in 1h 59m 07s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e3eb259316742fdbda9297ab829b490 : SUCCESS in 1h 50m 46s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/dbc2bb29ecc545a08d7a1b55fc098521 : SUCCESS in 2h 12m 54s\n- openstack-ansible-upgrade_2025.1-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/df468583a02244d692596544300511cf : SUCCESS in 2h 46m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e98f2273b26141979c483a0d82c35724 : SUCCESS in 2h 56m 31s\n- openstack-ansible-upgrade_2025.1-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4d34701bdd874585a7d7ab35a9680eeb : SUCCESS in 2h 34m 14s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99fe27d0110d4602a22bf35730250dfe : SUCCESS in 2h 41m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830212ba62df4897ac5784aa501f9b17 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":6},{"id":"2600a6bbea91d859b610cf6dc2141dceb8290b1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 23:54:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"77f38097448b201d3c9ef6261eb221e13cccd790","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 01:57:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8af8016f98c54bb6b5cda6bb3baedd62\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8467860b52d544c8adfbf577b4eb12e9 : SUCCESS in 13m 52s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f243cdc4474c439a83d50e3cdca10080 : SUCCESS in 1h 00m 39s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a6c26506a0f54b0f8e8c1f2cd0f2b130 : SUCCESS in 1h 38m 37s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/6d0cdb14f4ea404284f5c3464c32c0bb : SUCCESS in 1h 21m 26s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0327e22deab54771b85b663b0a2a8655 : SUCCESS in 1h 04m 49s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/23ac011e1d154a239a08c82c4e87f0ff : SUCCESS in 1h 00m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d0e5c454b02d42f3a5bd96cacaa6fe64 : SUCCESS in 1h 43m 11s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/709169d0840841a29906d92ee2f05e3e : SUCCESS in 1h 52m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ad950e0d15b406d8dadd330d67c61ce : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":6},{"id":"5e12d97e57e4785439629064d6361ec0a63c8038","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 01:57:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"1c300bb79018782ebfb6b7b4b505d3092ce073d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 01:59:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6293854e5387494b9d04257681984963\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84510d72bfe5405ea0f192c184f499c8 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cc27ccfdc3dc7663649f823ac38c1e6d21894d1a","revisions":{"248e355f56c92a10eeca52609d01cd45e0e15e14":{"kind":"REWORK","_number":1,"created":"2026-03-20 20:26:02.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/02/981602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/02/981602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/1"}}},"commit":{"parents":[{"commit":"1906235e9dcc91d6c1010c1e7def4f1533932dd5","subject":"Remove unused Makefile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/1906235e9dcc91d6c1010c1e7def4f1533932dd5"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:25:42.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:26:00.000000000","tz":60},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `placement_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nChange-Id: I62b8bdd569ba54e8d18450286234617ff1ce254e\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/248e355f56c92a10eeca52609d01cd45e0e15e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/248e355f56c92a10eeca52609d01cd45e0e15e14"}]},"branch":"refs/heads/master"},"6bca152a6eace76a66aa2256c123c9e3f66aae2e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-10 07:40:28.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"real_uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/02/981602/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/02/981602/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/2"}}},"commit":{"parents":[{"commit":"07b446d9ed15de96e5b76b20c8d83d762d5edf9e","subject":"Use Direct URL syntax for pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/07b446d9ed15de96e5b76b20c8d83d762d5edf9e"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:25:42.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-10 07:40:28.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `placement_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nChange-Id: I62b8bdd569ba54e8d18450286234617ff1ce254e\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6bca152a6eace76a66aa2256c123c9e3f66aae2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/6bca152a6eace76a66aa2256c123c9e3f66aae2e"}]},"branch":"refs/heads/master","description":"Rebase"},"d1fa489007e8125cd73b3f9173bcf3c60e7db432":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-15 08:59:10.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/02/981602/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/02/981602/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/3"}}},"commit":{"parents":[{"commit":"5a0b37b399883d822dd74c61776120b001917b62","subject":"Use Direct URL syntax for pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/5a0b37b399883d822dd74c61776120b001917b62"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:25:42.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-15 08:59:10.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `placement_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nChange-Id: I62b8bdd569ba54e8d18450286234617ff1ce254e\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d1fa489007e8125cd73b3f9173bcf3c60e7db432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/d1fa489007e8125cd73b3f9173bcf3c60e7db432"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"6bca152a6eace76a66aa2256c123c9e3f66aae2e","theirs":"5a0b37b399883d822dd74c61776120b001917b62","contains_conflicts":false}},"a1dbd32399058a9bea6c31b17e95aa85ea3629b0":{"kind":"REWORK","_number":4,"created":"2026-04-15 12:44:19.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/02/981602/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/02/981602/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/4"}}},"commit":{"parents":[{"commit":"5a0b37b399883d822dd74c61776120b001917b62","subject":"Use Direct URL syntax for pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/5a0b37b399883d822dd74c61776120b001917b62"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:25:42.000000000","tz":60},"committer":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-15 12:43:48.000000000","tz":120},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `placement_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nChange-Id: I62b8bdd569ba54e8d18450286234617ff1ce254e\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/a1dbd32399058a9bea6c31b17e95aa85ea3629b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/a1dbd32399058a9bea6c31b17e95aa85ea3629b0"}]},"branch":"refs/heads/master"},"4cf3cc1dfeb1772cdcff688a67675670be04033e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-04-18 10:47:37.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/02/981602/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/02/981602/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/5"}}},"commit":{"parents":[{"commit":"5a0b37b399883d822dd74c61776120b001917b62","subject":"Use Direct URL syntax for pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/5a0b37b399883d822dd74c61776120b001917b62"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:25:42.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-18 10:47:37.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `gnocchi_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nChange-Id: I62b8bdd569ba54e8d18450286234617ff1ce254e\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4cf3cc1dfeb1772cdcff688a67675670be04033e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/4cf3cc1dfeb1772cdcff688a67675670be04033e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"cc27ccfdc3dc7663649f823ac38c1e6d21894d1a":{"kind":"REWORK","_number":6,"created":"2026-04-18 10:48:20.000000000","uploader":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"ref":"refs/changes/02/981602/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_gnocchi","ref":"refs/changes/02/981602/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_gnocchi refs/changes/02/981602/6"}}},"commit":{"parents":[{"commit":"5a0b37b399883d822dd74c61776120b001917b62","subject":"Use Direct URL syntax for pip packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/5a0b37b399883d822dd74c61776120b001917b62"}]}],"author":{"name":"Damian Dabrowski","email":"damian.dabrowski@cleura.com","date":"2026-03-20 20:25:42.000000000","tz":60},"committer":{"name":"Damian Dąbrowski","email":"damian.dabrowski@cleura.com","date":"2026-04-18 10:48:14.000000000","tz":0},"subject":"Add hashi_vault pki backend support","message":"Add hashi_vault pki backend support\n\nThis patch defines necessary variables for the \u0027hashi_vault\u0027 pki backend\nwhich is the alternative to the default \u0027standalone\u0027 backend.\n\nAdditionally, it:\n- changes the format of `gnocchi_pki_san` to the new one changed in [1]\n- passes `pki_default_backend` when trigerring PKI role so its aware of\nthe default backend in case it\u0027s not explicitly specified in the cert\ndefinition.\n- adopts this role to the recent changes in PKI role [2][3][4]\n\n[1] https://review.opendev.org/c/openstack/openstack-ansible/+/948886\n[2] https://opendev.org/openstack/ansible-role-pki/commit/fc7db0207421dc97b484cf2ef04784292cc683f2\n[3] https://opendev.org/openstack/ansible-role-pki/commit/7cff89ee71ea99cbe1fd16f01f8341b3609b793b\n[4] https://opendev.org/openstack/ansible-role-pki/commit/f03bcc19d585f0e5fc67146015c44b9650b2fe13\n\nChange-Id: I62b8bdd569ba54e8d18450286234617ff1ce254e\nSigned-off-by: Damian Dabrowski \u003cdamian.dabrowski@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cc27ccfdc3dc7663649f823ac38c1e6d21894d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_gnocchi/commit/cc27ccfdc3dc7663649f823ac38c1e6d21894d1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
