)]}'
{"id":"openstack%2Fopenstack-ansible-os_heat~657993","triplet_id":"openstack%2Fopenstack-ansible-os_heat~master~I87aadbf0dd8c4233e93f533468029d973e414556","project":"openstack/openstack-ansible-os_heat","branch":"master","hashtags":[],"change_id":"I87aadbf0dd8c4233e93f533468029d973e414556","subject":"Replace openstack.org URLs with opendev.org URLs","status":"ABANDONED","created":"2019-05-09 11:06:32.000000000","updated":"2022-05-10 16:40:14.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c24e5994322d010b4073b26101f165c0ede850a1","_number":657993,"virtual_id_number":657993,"owner":{"_account_id":30387,"name":"weidong.huang","email":"huang.weidong@99cloud.net","username":"huang.weidong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-15 02:14:17.000000000","_account_id":30387,"name":"weidong.huang","email":"huang.weidong@99cloud.net","username":"huang.weidong"},{"value":1,"date":"2019-05-09 12:38:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30387,"name":"weidong.huang","email":"huang.weidong@99cloud.net","username":"huang.weidong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-05-09 13:00:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30387,"name":"weidong.huang","email":"huang.weidong@99cloud.net","username":"huang.weidong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30387,"name":"weidong.huang","email":"huang.weidong@99cloud.net","username":"huang.weidong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":30387,"name":"weidong.huang","email":"huang.weidong@99cloud.net","username":"huang.weidong"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:38:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-09 13:00:14.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"}],"messages":[{"id":"e01dd8b756cc86fbfc554478ca10df4c2ff26d27","author":{"_account_id":30387,"name":"weidong.huang","email":"huang.weidong@99cloud.net","username":"huang.weidong"},"date":"2019-05-09 11:06:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3ca791e9dec4b67c0b1d65d7e471314bbefe14c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 12:38:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/93/657993/1/check/openstack-ansible-linters/080585f/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/93/657993/1/check/openstack-tox-docs/ba28cf2/html/ : SUCCESS in 2m 29s\n- build-openstack-releasenotes http://logs.openstack.org/93/657993/1/check/build-openstack-releasenotes/c9b238e/html/ : SUCCESS in 2m 28s\n- openstack-ansible-deploy-aio_metal_heat-centos-7 http://logs.openstack.org/93/657993/1/check/openstack-ansible-deploy-aio_metal_heat-centos-7/6a252c1/ : SUCCESS in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal_heat-debian-stable http://logs.openstack.org/93/657993/1/check/openstack-ansible-deploy-aio_metal_heat-debian-stable/a86b6ce/ : SUCCESS in 54m 57s\n- openstack-ansible-deploy-aio_metal_heat-ubuntu-bionic http://logs.openstack.org/93/657993/1/check/openstack-ansible-deploy-aio_metal_heat-ubuntu-bionic/be904e9/ : SUCCESS in 58m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ad29f2d8b8729f29267c865c7b03a479e44046e4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-09 13:00:14.000000000","message":"Patch Set 1: Code-Review-1\n\nA lot of these files are synchronised automatically from openstack-ansible-tests using the proposal bot. They should not be modified in place like this.","accounts_in_message":[],"_revision_number":1},{"id":"f751af95422fc236fded2018bbf8e4e119a5ffdb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-09 13:00:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1661d4bbaba74bf55b42923968b7adb9030d770e","author":{"_account_id":30387,"name":"weidong.huang","email":"huang.weidong@99cloud.net","username":"huang.weidong"},"date":"2019-05-15 02:14:17.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nAbandoned","accounts_in_message":[],"_revision_number":1},{"id":"c24e5994322d010b4073b26101f165c0ede850a1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-05-10 16:40:14.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bf13a3d9da303652a43e73ad13776047b713bb6","revisions":{"2bf13a3d9da303652a43e73ad13776047b713bb6":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:06:32.000000000","uploader":{"_account_id":30387,"name":"weidong.huang","email":"huang.weidong@99cloud.net","username":"huang.weidong"},"ref":"refs/changes/93/657993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/93/657993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/93/657993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/93/657993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/93/657993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/93/657993/1"}}},"commit":{"parents":[{"commit":"4d74a2c4a2d6ac2a86799cfe1bde4faedf61c5ed","subject":"Merge \"missing document start \"---\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/4d74a2c4a2d6ac2a86799cfe1bde4faedf61c5ed"}]}],"author":{"name":"huang.weidong","email":"huang.weidong@99cloud.net","date":"2019-05-09 11:04:45.000000000","tz":480},"committer":{"name":"huang.weidong","email":"huang.weidong@99cloud.net","date":"2019-05-09 11:06:27.000000000","tz":480},"subject":"Replace openstack.org URLs with opendev.org URLs","message":"Replace openstack.org URLs with opendev.org URLs\n\nChange-Id: I87aadbf0dd8c4233e93f533468029d973e414556\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/2bf13a3d9da303652a43e73ad13776047b713bb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/2bf13a3d9da303652a43e73ad13776047b713bb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
