)]}'
{"id":"openstack%2Fopenstack-ansible-os_heat~658596","triplet_id":"openstack%2Fopenstack-ansible-os_heat~stable%2Fstein~I12d1ecbfb7594c48110255714b579be35bdb1854","project":"openstack/openstack-ansible-os_heat","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I12d1ecbfb7594c48110255714b579be35bdb1854","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-11 23:58:44.000000000","updated":"2019-05-12 03:14:41.000000000","submitted":"2019-05-12 03:14:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658596-1557630881198-1a85bf1c","meta_rev_id":"ae7dfeab41fd9791772a16089cd7d18e762fb6c7","_number":658596,"virtual_id_number":658596,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-12 03:14:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-12 01:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-12 01:54:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-12 01:54:49.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-12 03:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"631b2fd1ec056a1036ac970dba90745edf184f3d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-11 23:58:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4679a51b2147411e54406b0e5a6bae2a3f0b3048","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:26:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/96/658596/1/check/openstack-ansible-linters/7f71a65/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/96/658596/1/check/openstack-tox-docs/68560e3/html/ : SUCCESS in 2m 41s\n- build-openstack-releasenotes http://logs.openstack.org/96/658596/1/check/build-openstack-releasenotes/47e5794/html/ : SUCCESS in 2m 33s\n- openstack-ansible-deploy-aio_metal_heat-centos-7 http://logs.openstack.org/96/658596/1/check/openstack-ansible-deploy-aio_metal_heat-centos-7/21911dc/ : SUCCESS in 1h 16m 09s\n- openstack-ansible-deploy-aio_metal_heat-debian-stable http://logs.openstack.org/96/658596/1/check/openstack-ansible-deploy-aio_metal_heat-debian-stable/2c745c7/ : SUCCESS in 57m 45s\n- openstack-ansible-deploy-aio_metal_heat-ubuntu-bionic http://logs.openstack.org/96/658596/1/check/openstack-ansible-deploy-aio_metal_heat-ubuntu-bionic/e5ea158/ : SUCCESS in 56m 29s","accounts_in_message":[],"_revision_number":1},{"id":"91f4921fe0f9ad9c19ba2144bd275735e0a688af","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-12 01:54:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b92c0f6e80993eb704eb249801404089a47cdc49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 01:55:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d5930f5bdca615a0d2c576ae05ae11f18de4542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:14:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/96/658596/1/gate/openstack-ansible-linters/95d722a/ : SUCCESS in 6m 25s\n- openstack-tox-docs http://logs.openstack.org/96/658596/1/gate/openstack-tox-docs/3bbbb66/html/ : SUCCESS in 3m 40s\n- build-openstack-releasenotes http://logs.openstack.org/96/658596/1/gate/build-openstack-releasenotes/7468eb9/html/ : SUCCESS in 3m 14s\n- openstack-ansible-deploy-aio_metal_heat-centos-7 http://logs.openstack.org/96/658596/1/gate/openstack-ansible-deploy-aio_metal_heat-centos-7/192dff5/ : SUCCESS in 1h 17m 10s\n- openstack-ansible-deploy-aio_metal_heat-debian-stable http://logs.openstack.org/96/658596/1/gate/openstack-ansible-deploy-aio_metal_heat-debian-stable/7b57f5a/ : SUCCESS in 57m 51s\n- openstack-ansible-deploy-aio_metal_heat-ubuntu-bionic http://logs.openstack.org/96/658596/1/gate/openstack-ansible-deploy-aio_metal_heat-ubuntu-bionic/c0ff5b2/ : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e48f4583489b7802b6a9392b0bf9e21cea5a98df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:14:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56029e5daaf7337b953ac4a9eae3a73cd7a052c8","revisions":{"56029e5daaf7337b953ac4a9eae3a73cd7a052c8":{"kind":"REWORK","_number":1,"created":"2019-05-11 23:58:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/96/658596/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/96/658596/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/96/658596/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/96/658596/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/96/658596/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/96/658596/1"}}},"commit":{"parents":[{"commit":"197f116db2202150936e53c4f112b5c2cf9fe00e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/197f116db2202150936e53c4f112b5c2cf9fe00e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-11 23:58:43.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I12d1ecbfb7594c48110255714b579be35bdb1854\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/56029e5daaf7337b953ac4a9eae3a73cd7a052c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/56029e5daaf7337b953ac4a9eae3a73cd7a052c8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
