)]}'
{"id":"openstack%2Fopenstack-ansible-os_heat~680045","triplet_id":"openstack%2Fopenstack-ansible-os_heat~master~Ife7537490972f3f9ff05fc4b44595a023ffb7e4c","project":"openstack/openstack-ansible-os_heat","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ife7537490972f3f9ff05fc4b44595a023ffb7e4c","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-09-04 15:14:17.000000000","updated":"2019-09-06 10:42:19.000000000","submitted":"2019-09-06 10:39:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680045-1567766382999-7e60562c","meta_rev_id":"b56a24c87091f6ac98817354751d605a6b758c6f","_number":680045,"virtual_id_number":680045,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-09-06 10:39:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-09-05 21:52:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-09-05 21:52:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 21:52:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-09-06 10:39:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"127e8a3bfaa9a061cfb2fcc49c70c22787ffa473","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-09-04 15:14:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a72544586a4c2c92f30aa2ba4892fce6ef228525","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 18:09:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1652b188dc9f41b6a3d6c61495268a5a : POST_FAILURE in 8m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3fa87b922e604c3eb05b15166741cab5 : POST_FAILURE in 1h 12m 56s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/326e525f724a4d4c9ef228762a664337 : FAILURE in 35m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/240a6330ccfe4ce9a0f92e79b0ee9918 : SUCCESS in 1h 01m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c738b57943a2451380b97f806004dc47 : SUCCESS in 1h 09m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aa4f577498c147cc8ecd334a52198faa : SUCCESS in 1h 46m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cde5f3efb7194b9d86689a98aa17e9a3 : POST_FAILURE in 35m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4076537daeb6488cbe6389baa74e2fcd : POST_FAILURE in 1h 03m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbb21c9b62746ee83ffcc8b51ba8ec9 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2b5be483750e230d269c756e1971bfbc344ba7b6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 14:47:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"36d2ccf3423474f3702bf544a9343bc4255f3e14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 18:55:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bdbf9c31718148859dea0d9daf643cc4 : SUCCESS in 6m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/62acd59bd9a945e0b909b7220e065f85 : SUCCESS in 1h 03m 02s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5fe92b86508e47a9910b05ad7b590e9a : FAILURE in 34m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43d58ea7611a467b892007febb622ffb : SUCCESS in 56m 28s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/21c256c9c931479fac106ad038ba12c5 : SUCCESS in 1h 36m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/65d1b2197c5648b5b63a456c8acf8ee7 : SUCCESS in 1h 48m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ff9fe962cf0245eb8738efc5e1a546f4 : SUCCESS in 1h 20m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a78079571824a9ba32723c44369fbe5 : SUCCESS in 1h 09m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9668c3de61224fd9be2fa73adc83d597 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b03053c614766d1b229ed9015161af6a5cb938ea","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-05 21:52:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1fbb9bb06a0493ef14b51d6555256dca05b10588","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 21:53:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e836396c2381f7803699a3246e24c5c46f751cbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 23:36:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/72763896d64d44968675848f97d5d094 : SUCCESS in 6m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d6b5e3a95cc04e0f8080bc71cccd7d5e : SUCCESS in 1h 09m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbd8d417d7b649f1a5096330a23c67f9 : FAILURE in 24m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7acfbaceff6c40f28f2131095a766f01 : SUCCESS in 1h 19m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/74c66e676edd410fb410575c1cc51806 : SUCCESS in 1h 38m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/238c0c3355524216aabc4e21d3b7a4bf : SUCCESS in 1h 18m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d4b580240d1426da8218b980b07bdf5 : SUCCESS in 1h 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a088b298fe4cf48f29d2c706c3beb9 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a88b02dd30eddb28f95ae2180854392b0db38a12","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-06 07:08:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0f8e4f4f1907776d27880d37d86b3a000a3899ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 08:54:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f3c48ab45a0a4f618eb17e7ca823372b : SUCCESS in 5m 48s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ebbc62a1cc68485eb57f044918235310 : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/690da7eaf12b44de9b7798c127bc2318 : FAILURE in 30m 33s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/802df0e40ac3423ca00f58b4dfa8db3a : SUCCESS in 52m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b47fc11d8dda43f59311acceb95a068b : SUCCESS in 1h 10m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f1b439d0136247eda55be3e8a3a9d4e6 : SUCCESS in 1h 36m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/91aacf88870b44fda3e14f0de786f78b : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44478f2f935d47f7be53c7893f563bfa : SUCCESS in 1h 06m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c0805a9ea649bc89a2a59918596c41 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0f681b16c8b3c585816180a6d9d536a1ffc0e680","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 08:54:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a12292e31153ae0071d5cf7ce77e17d258669171","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:39:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bff9a57bbf6240279bb8196f8126b12d : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/054b47b173a24f03be16fcfaef7595ff : SUCCESS in 1h 07m 38s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f4618cc6e6b4db982f193395e374bba : SUCCESS in 53m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/25948984c1504980b7ae8fbed052ab34 : SUCCESS in 1h 09m 07s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9c1bcc6d39f340d4bd4906860691a62d : SUCCESS in 1h 34m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/961e29982a964c908656790a1c97091a : SUCCESS in 1h 16m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72289c0baead45228435ec4b46eed5c5 : SUCCESS in 1h 04m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3c4cee9c2644efaf55de261ca3d61f : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"dfc79b8df45baa6164845101b5ac6dded0a4b0de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:39:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f96de02aeba8d478c3230302b6bd288f04eb8911","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:42:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7fc22f357ed421482f526f049f90926 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b7002e8cae9e10f3fb3d2de7a66fc0b62f17523","revisions":{"2b7002e8cae9e10f3fb3d2de7a66fc0b62f17523":{"kind":"REWORK","_number":1,"created":"2019-09-04 15:14:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/680045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/45/680045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/45/680045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/45/680045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/45/680045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/45/680045/1"}}},"commit":{"parents":[{"commit":"288634ce0bf042bed614b3f764753d7b65a7170f","subject":"Fix keystone endpoint for heat servers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/288634ce0bf042bed614b3f764753d7b65a7170f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 15:14:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-09-04 15:14:13.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ife7537490972f3f9ff05fc4b44595a023ffb7e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/2b7002e8cae9e10f3fb3d2de7a66fc0b62f17523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/2b7002e8cae9e10f3fb3d2de7a66fc0b62f17523"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
