)]}'
{"id":"openstack%2Fopenstack-ansible-os_heat~745186","triplet_id":"openstack%2Fopenstack-ansible-os_heat~master~I1485160f330fa289e02874e9286cf7a27009ea76","project":"openstack/openstack-ansible-os_heat","branch":"master","hashtags":[],"change_id":"I1485160f330fa289e02874e9286cf7a27009ea76","subject":"Simplify service creation","status":"MERGED","created":"2020-08-06 16:42:11.000000000","updated":"2020-08-12 12:54:29.000000000","submitted":"2020-08-12 12:51:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745186-1597236663952-480d4010","meta_rev_id":"88f4ffb2f1039bb8665e7b26ccfa5f0c14e3cd6b","_number":745186,"virtual_id_number":745186,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-12 12:51:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-08-12 10:59:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-08 10:37:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-08-12 10:59:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-08 10:37:38.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-12 10:59:31.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-08-12 12:51:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa7d46340d93cb998b88d84489b0b17662bbc88f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-06 16:42:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5fcc59a8db110bbf0cdbf3cae5d5e94f291ea015","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-06 16:42:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5ce659d180b37601ead6289955c79d90435030fe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-06 16:47:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a60e2066d157e899bab0bf7edffa076895d15787","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-06 16:51:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1cc59a8b47207e5c3a613c8022ab82fdd7000456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 19:52:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ef7cd21d01194d988c19c0f5acf6ed8d : SUCCESS in 8m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bb28bdd609334cf898f88790ea395bfe : FAILURE in 1h 04m 46s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cedc59620da94bec98545f7d0aa3818d : FAILURE in 50m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96760a790fea4ae390015510724f4c7d : FAILURE in 55m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e289e05e5d5479e83bd68f0a2e79f3c : FAILURE in 41m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f39d3da4f5744daa97c29585750132ed : FAILURE in 50m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6cebf39bdd644616909dc48380bc57c7 : FAILURE in 1h 07m 08s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ec3a330cc47d49f98918424a931e53d2 : FAILURE in 1h 08m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8f7ebf2690c8446ab9e489bb217d758e : FAILURE in 1h 02m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/499d4c4c48a94e58a436d34153a86332 : FAILURE in 59m 41s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6199c57223ba44ee81943534c4c0aea7 : SUCCESS in 2h 55m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63fd18cfab06478091774de98b6587af : SUCCESS in 2h 02m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ca025d75abb44f8960c530c4057ba11 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":4},{"id":"9a9d02fc9cbeab6bd97cc1d809fb908a5ab3f7f3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-06 19:59:41.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"e867e40f7a5e10178775eaa833ef83de9ed7910e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-06 20:28:32.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"10b392d9601422545b0964f7760b934c3b9ba007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 23:22:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f6efba6865b249379e9fcfba51af472f : SUCCESS in 8m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/66eca20eb4784c5599ddfc7a00fd0621 : SUCCESS in 1h 07m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7f43222e68d54a1c970bd0f3674ddca6 : SUCCESS in 1h 07m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2358a11b31c149f2ad2f604f9b5dab75 : SUCCESS in 59m 35s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/257d05ce3fb34190a0447ebd6fe8c280 : SUCCESS in 1h 02m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1b45848c69014d02813a67e241a638c6 : SUCCESS in 1h 13m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/99eda44882b340af9306c15f58d33b73 : SUCCESS in 1h 25m 48s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ce2c48c223674a59955219875e967ff8 : SUCCESS in 1h 26m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d6ef29622b046f7a1254317f0659e68 : SUCCESS in 1h 17m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e3008865811412db3e0a919031bc58f : SUCCESS in 1h 08m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7968dff72b5541d4891307abc8d5dbf2 : SUCCESS in 2h 43m 52s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57de9e5dfafd488e8f6de1126035c6ac : SUCCESS in 1h 58m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee342209e7c24772aadc97a670089ce2 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":6},{"id":"f5d4a5f6bb80185a4c571984eb5b5399fa4ae840","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-08 10:37:38.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"2aad02fd64b694a41a3170cd8ad574ab9399a2c1","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-08-12 10:59:31.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"baf3bbe09a4e183dd86b3b5bce6597a278fe1a21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 10:59:42.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c673e103804551c8a594df903344ccc7de873b96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 12:51:03.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8e281355b9754381a1f060912797a21f : SUCCESS in 8m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/730cde87098a49a2abf2f72edfc8b870 : SUCCESS in 1h 37m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d708adb5f0904c55b595c4c23e181429 : SUCCESS in 1h 32m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fac70cc2bacc4fa395e93fec5719b9ba : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52f88cfb48ca40c298e6cf30664dd92e : SUCCESS in 53m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d8f2f3173a4445649e91bd2ce33420f7 : SUCCESS in 1h 05m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d567ad51e8d47dc9f57c70e3b0c8a39 : SUCCESS in 1h 20m 55s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/96ddc5b670ba4e0d8624ccefd00f5385 : SUCCESS in 1h 18m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daa85abb82d64cd6bfa321db485ad4ad : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6249d5822664acea00f2518f20145e6 : SUCCESS in 1h 09m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d004a1d41e9406fbf20e4649b54d623 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":6},{"id":"aa2655bd03c92a69de2dd68c1e57756b63fd0099","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 12:51:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"cee438aa7c2e68dfffdce6c67c8c9baaac1d41cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 12:52:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df412ecbc5a240dda470603a9f11f6da : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":6},{"id":"88f4ffb2f1039bb8665e7b26ccfa5f0c14e3cd6b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-12 12:54:29.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 037521a3c5051cd0a5534625d7dd38384ee89aa6","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"2878de0578f9a03cd9c9d0730c40693762a32f64","revisions":{"212c4969fd771a8d8534813b39ce45862fab11b7":{"kind":"REWORK","_number":1,"created":"2020-08-06 16:42:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/745186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/86/745186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/1"}}},"commit":{"parents":[{"commit":"56fdb1837714fe14c41af40ed560db306b20500d","subject":"Fix _project_domain for heat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/56fdb1837714fe14c41af40ed560db306b20500d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:41:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:41:57.000000000","tz":180},"subject":"Simplify service creation","message":"Simplify service creation\n\nSince ansible feature has been implemented and we can reference\nopenstack domain with it\u0027s name but not only id,\nwe can simplify service creation task.\n\nChange-Id: I1485160f330fa289e02874e9286cf7a27009ea76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/212c4969fd771a8d8534813b39ce45862fab11b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/212c4969fd771a8d8534813b39ce45862fab11b7"}]},"branch":"refs/heads/master"},"b1b24a9a4921717c29e7f3b50f621a68a7f8424c":{"kind":"NO_CHANGE","_number":2,"created":"2020-08-06 16:42:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/745186/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/86/745186/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/2"}}},"commit":{"parents":[{"commit":"d3b9eb8e6af95aa04ab4b6b4a6fc56a311087f13","subject":"Fix _project_domain for heat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/d3b9eb8e6af95aa04ab4b6b4a6fc56a311087f13"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:41:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:42:57.000000000","tz":0},"subject":"Simplify service creation","message":"Simplify service creation\n\nSince ansible feature has been implemented and we can reference\nopenstack domain with it\u0027s name but not only id,\nwe can simplify service creation task.\n\nChange-Id: I1485160f330fa289e02874e9286cf7a27009ea76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/b1b24a9a4921717c29e7f3b50f621a68a7f8424c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/b1b24a9a4921717c29e7f3b50f621a68a7f8424c"}]},"branch":"refs/heads/master"},"3135635bcfdf053f95cb4c30e8998a61e333fa52":{"kind":"REWORK","_number":3,"created":"2020-08-06 16:47:58.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/745186/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/86/745186/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/3"}}},"commit":{"parents":[{"commit":"1fec9645a9ae1475df468b2ce9af2a22272e6c0e","subject":"Fix _project_domain for heat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/1fec9645a9ae1475df468b2ce9af2a22272e6c0e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:41:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:47:42.000000000","tz":180},"subject":"Simplify service creation","message":"Simplify service creation\n\nSince ansible feature has been implemented and we can reference\nopenstack domain with it\u0027s name but not only id,\nwe can simplify service creation task.\n\nChange-Id: I1485160f330fa289e02874e9286cf7a27009ea76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/3135635bcfdf053f95cb4c30e8998a61e333fa52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/3135635bcfdf053f95cb4c30e8998a61e333fa52"}]},"branch":"refs/heads/master"},"ad84b4e97b4b5d92426873d6ec7f4ad7e4d9acec":{"kind":"REWORK","_number":4,"created":"2020-08-06 16:51:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/745186/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/86/745186/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/4"}}},"commit":{"parents":[{"commit":"1fec9645a9ae1475df468b2ce9af2a22272e6c0e","subject":"Fix _project_domain for heat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/1fec9645a9ae1475df468b2ce9af2a22272e6c0e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:41:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:51:30.000000000","tz":180},"subject":"Simplify service creation","message":"Simplify service creation\n\nSince ansible feature has been implemented and we can reference\nopenstack domain with it\u0027s name but not only id,\nwe can simplify service creation task.\n\nChange-Id: I1485160f330fa289e02874e9286cf7a27009ea76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/ad84b4e97b4b5d92426873d6ec7f4ad7e4d9acec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/ad84b4e97b4b5d92426873d6ec7f4ad7e4d9acec"}]},"branch":"refs/heads/master"},"14fd023e88e7cc94317a3c1669a335f41137ce53":{"kind":"REWORK","_number":5,"created":"2020-08-06 19:59:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/745186/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/86/745186/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/5"}}},"commit":{"parents":[{"commit":"780dc2eb597d3736b702d5f8cdf830e83d177e3d","subject":"Fix _project_domain for heat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/780dc2eb597d3736b702d5f8cdf830e83d177e3d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:41:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 19:59:28.000000000","tz":180},"subject":"Simplify service creation","message":"Simplify service creation\n\nSince ansible feature has been implemented and we can reference\nopenstack domain with it\u0027s name but not only id,\nwe can simplify service creation task.\n\nChange-Id: I1485160f330fa289e02874e9286cf7a27009ea76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/14fd023e88e7cc94317a3c1669a335f41137ce53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/14fd023e88e7cc94317a3c1669a335f41137ce53"}]},"branch":"refs/heads/master"},"2878de0578f9a03cd9c9d0730c40693762a32f64":{"kind":"REWORK","_number":6,"created":"2020-08-06 20:28:32.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/86/745186/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/86/745186/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/86/745186/6"}}},"commit":{"parents":[{"commit":"780dc2eb597d3736b702d5f8cdf830e83d177e3d","subject":"Fix _project_domain for heat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/780dc2eb597d3736b702d5f8cdf830e83d177e3d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-06 16:41:57.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-08-06 20:28:29.000000000","tz":0},"subject":"Simplify service creation","message":"Simplify service creation\n\nSince ansible feature has been implemented and we can reference\nopenstack domain with it\u0027s name but not only id,\nwe can simplify service creation task.\n\nChange-Id: I1485160f330fa289e02874e9286cf7a27009ea76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/2878de0578f9a03cd9c9d0730c40693762a32f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/2878de0578f9a03cd9c9d0730c40693762a32f64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
