)]}'
{"id":"openstack%2Fopenstack-ansible-os_heat~755565","triplet_id":"openstack%2Fopenstack-ansible-os_heat~master~Ib96a0ec7795b258548ecc3dd1baaa7c3cd35ab3f","project":"openstack/openstack-ansible-os_heat","branch":"master","topic":"osa-linters-update","hashtags":[],"change_id":"Ib96a0ec7795b258548ecc3dd1baaa7c3cd35ab3f","subject":"Fix linter errors","status":"MERGED","created":"2020-10-01 15:45:31.000000000","updated":"2020-10-01 17:18:46.000000000","submitted":"2020-10-01 17:13:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755565-1601572406932-8f1aa1ae","meta_rev_id":"f73d675705516894344d153da4317d5ee8521c87","_number":755565,"virtual_id_number":755565,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 17:13:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-10-01 16:54:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-10-01 16:54:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 16:54:00.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-01 17:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e4ff32e6ed9ea16a95544e77866b99f02c4d0397","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-01 15:45:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7dd5d3643b6af18175556d46cdeadda78e6b591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:00:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/574b5c71e3a042d4a6faaa370ecc27b6 : FAILURE in 7m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01c09097c99346a7ad1526de9f14f0d9 : FAILURE in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"07099458abe61397ad1fd9c42999f23ffcdf64a8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-01 16:02:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"30bd7db13642e882468da96fd5beb698c8edc7c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:20:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/105e4d61b2e54462b1a97e76d7070a1a : FAILURE in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a324aeb281c14d528ba7d4fa8fcee7e2 : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"cb37836aa6baaf3b77a0c55316567f0cb1b67203","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-01 16:23:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0b502d79cb8567571a857f5576f4253e427f989b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-01 16:33:15.000000000","message":"Topic set to osa-linters-update","accounts_in_message":[],"_revision_number":3},{"id":"77b8c8655791b62d880802f62f2395ea499dbb40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:47:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/03a38b8d5e5741e48958e8f0773467ca : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f5b5ca223d64d1daa0acfba3690eb2f : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"19a924ee194cdf13ebbe5d3e80325e59c5f631fb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-01 16:54:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f86677596689211970e3437f0634eba1a93188f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 16:54:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"20f83274fab595f23175a81d51f2d272a91395b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 17:13:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f506b8c98912edbabc5814deba6a3ab5568ec2ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 17:13:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/51d0e43e8031447981b9d0b84c28dbc9 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69479bd3197840b79540f2be14dff8ac : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":3},{"id":"f73d675705516894344d153da4317d5ee8521c87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 17:18:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db07904c0e44c7587b7c6a5eed48820 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"22cef2957bf52c6ee2d0596ebd07ba6c04169cb8","revisions":{"9e7c6afe37b47920c73cb5a838300691231f2847":{"kind":"REWORK","_number":1,"created":"2020-10-01 15:45:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/65/755565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/65/755565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/1"}}},"commit":{"parents":[{"commit":"8c585989ab6cc0929e2e424bc7ffa8673e3031a0","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/8c585989ab6cc0929e2e424bc7ffa8673e3031a0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:45:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:45:19.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Ib96a0ec7795b258548ecc3dd1baaa7c3cd35ab3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/9e7c6afe37b47920c73cb5a838300691231f2847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/9e7c6afe37b47920c73cb5a838300691231f2847"}]},"branch":"refs/heads/master"},"7a9ecccf35dcc9f1de091c793e8bf4e55cea61b1":{"kind":"REWORK","_number":2,"created":"2020-10-01 16:02:22.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/65/755565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/65/755565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/2"}}},"commit":{"parents":[{"commit":"8c585989ab6cc0929e2e424bc7ffa8673e3031a0","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/8c585989ab6cc0929e2e424bc7ffa8673e3031a0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:45:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 16:02:11.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Ib96a0ec7795b258548ecc3dd1baaa7c3cd35ab3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/7a9ecccf35dcc9f1de091c793e8bf4e55cea61b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/7a9ecccf35dcc9f1de091c793e8bf4e55cea61b1"}]},"branch":"refs/heads/master"},"22cef2957bf52c6ee2d0596ebd07ba6c04169cb8":{"kind":"REWORK","_number":3,"created":"2020-10-01 16:23:51.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/65/755565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/65/755565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/65/755565/3"}}},"commit":{"parents":[{"commit":"8c585989ab6cc0929e2e424bc7ffa8673e3031a0","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/8c585989ab6cc0929e2e424bc7ffa8673e3031a0"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 15:45:19.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-10-01 16:23:36.000000000","tz":60},"subject":"Fix linter errors","message":"Fix linter errors\n\nChange-Id: Ib96a0ec7795b258548ecc3dd1baaa7c3cd35ab3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/22cef2957bf52c6ee2d0596ebd07ba6c04169cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/22cef2957bf52c6ee2d0596ebd07ba6c04169cb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
