)]}'
{"id":"openstack%2Fopenstack-ansible-os_heat~985385","triplet_id":"openstack%2Fopenstack-ansible-os_heat~stable%2F2025.1~Ib36e9adf6f27fba662fed2dd955f586bde7b4c2d","project":"openstack/openstack-ansible-os_heat","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"31542":{"account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"last_update":"2026-04-22 09:49:57.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-21 17:02:14.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-22 11:45:52.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2026-04-22 11:45:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib36e9adf6f27fba662fed2dd955f586bde7b4c2d","subject":"Add keystoneauth options to [ec2authtoken]","status":"MERGED","created":"2026-04-20 12:03:51.000000000","updated":"2026-04-22 11:46:51.000000000","submitted":"2026-04-22 11:45:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"985385","cherry_pick_of_change":985382,"cherry_pick_of_patch_set":1,"meta_rev_id":"5130f787e66ab33c911828de292ebf0e725ef2d0","_number":985385,"virtual_id_number":985385,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-22 11:45:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-04-22 09:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-04-21 17:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2026-04-22 09:49:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-20 12:03:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2026-04-20 15:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 13:36:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-04-21 13:36:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2026-04-21 13:36:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"c549cdf8c36cdf255c7e5ee5b6a6a17d542112b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-20 12:03:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.2.","accounts_in_message":[],"_revision_number":1},{"id":"bf777489612b6fcc5e71c504c333a784159a7630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-20 15:38:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cb7fb8779f343c0b4c2c8dd772c5290\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d64c8c11f25458ba8da77d5e0af44a1 : SUCCESS in 21m 35s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/8e21156c38604ca1a90cbb72287513ca : SUCCESS in 1h 16m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3d875cc6911d479cb62048022e5430fe : FAILURE in 12m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/157d0f45a19d489ab2df126476fb58d2 : SUCCESS in 1h 14m 04s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/5b89ffe7fb8d4f438ad98a746f1faa38 : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/e36c603848b24de280282582706ad2f4 : SUCCESS in 1h 35m 45s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/3255a87a5a254956962f63ccd6e8935b : SUCCESS in 1h 41m 59s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3eb09fd87ab0438a8710210a29741208 : SUCCESS in 1h 24m 36s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/f2231c43e9984b40ae45be12463a132b : SUCCESS in 1h 20m 13s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/95834d0671ab498b870eb93b3ffdfcad : SUCCESS in 1h 35m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c14dfd1f72e5401089861293e641fb7c : SUCCESS in 1h 44m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28859c7c195d4985959b58777bbb66c6 : SUCCESS in 1h 42m 11s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/895a7017c87b4137af320ca00ce55a15 : SUCCESS in 1h 40m 12s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/145de3686e8c4422b90b385fbac32897 : SUCCESS in 1h 29m 21s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/28d8e5cb9fe24cb09bd13d4bce77cb5e : SUCCESS in 2h 06m 41s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/57476014994048c78f6ab318ad4de47d : SUCCESS in 2h 47m 34s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dcc03f60d4984f60bffd0081be2bc8e0 : SUCCESS in 2h 41m 04s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bedadcd4f054463f97e36e8ab8107c68 : FAILURE in 1h 02m 41s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50e49590d7554136a47453af974d7678 : SUCCESS in 2h 09m 33s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/dba82c59630c4463af2e95c7760737e3 : SUCCESS in 2h 34m 45s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73309a9010884b8a94e982d8cc3eb5b6 : SUCCESS in 2h 31m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db36e8a51044f2cb0ca8a30ad0938dc : SUCCESS in 8m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4befec3ffd15216da7b630e6add205352d490da9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-21 10:11:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6875ee13711901042f229cd87851c8e33af3e6d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-21 13:08:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07afdf0894d34f7abf69faf21573e4f2\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1ddbb2c5002e4bff9c9223f999297266 : SUCCESS in 10m 14s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9dcf0738c5ac4a43bfa8f733e8383417 : SUCCESS in 41m 27s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/38f88733444545a5a8fc84600abfb527 : FAILURE in 14m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e12c4a010f4949b6a81c44290a7e3709 : SUCCESS in 1h 10m 03s (non-voting)\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/221139b9115344a185d29cc2fd532be5 : SUCCESS in 51m 02s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/b7c172468dad414387068b77560bb5c5 : SUCCESS in 1h 27m 23s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b7a076be5d6450a80ea2300338bc9e9 : SUCCESS in 1h 32m 13s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/263b2ff7e47c4b018e5fb7a00eacad69 : SUCCESS in 1h 46m 16s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/aacf1e3f67d4414e9ff83ed8ce53644e : SUCCESS in 56m 01s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/66df0b9db43949f196ee5bbe77a0676d : SUCCESS in 57m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1eac6d2561fe46e589dd6881f1e73a45 : SUCCESS in 55m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e5befa550e849f499582807665d0246 : SUCCESS in 1h 41m 03s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/56b3c94a985a4efaadd98aacf05f844c : SUCCESS in 1h 35m 50s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a4595be65904177bc689c31ff1b8a49 : SUCCESS in 1h 27m 45s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80bc562a857949a686c1cac5df71a174 : SUCCESS in 54m 40s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a4c2050e614a417b9b9e894677539d67 : SUCCESS in 2h 27m 34s\n- openstack-ansible-upgrade_2024.1-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/382c6110433140eb94232bd9d6042514 : SUCCESS in 2h 22m 39s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97726f49c02d40be86761bd93a00d1a5 : SUCCESS in 1h 14m 13s\n- openstack-ansible-upgrade_2024.1-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3e6135cf3d946c5bf295224e66a635f : SUCCESS in 1h 16m 34s\n- openstack-ansible-upgrade-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e452a0306c6743929d440f6ddaa2d5ce : SUCCESS in 1h 12m 43s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77233d21e12948b591d00c18294587d0 : SUCCESS in 2h 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecccc29603fb4a9abd3e84376fc0a4cb : SUCCESS in 7m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a5f05d66ab8bb7569471400dcc0393df9eea410e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-21 17:02:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf3e44b7617c18f7f125a788b0fe7c94b62a3cf6","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-04-22 09:49:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d75ba2e0fdfbc7e16134473e6f88203f8376ee94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 09:53:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b6e7263d90c017ddfaad474daca867489215ade","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 11:45:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84c611be6afd441380d5e3364a20fae1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/264f2f8185c64b59b4999b920c771607 : SUCCESS in 13m 45s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3aa66bb975d24164b7e0ecd54969f29f : SUCCESS in 50m 28s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/8a36c3e31b30450abc20ef8dc3c6fabc : SUCCESS in 1h 26m 24s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3b3cde56412c4ab7a4e819a497b8c4ac : SUCCESS in 1h 30m 52s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/de9ad1cd126a47ef8b997e99a5db01cb : SUCCESS in 50m 18s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1a77d4bcb449413cab43550e5d26083a : SUCCESS in 51m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02901c085c3046d2ae01ef6446808c34 : SUCCESS in 1h 26m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/960907c75b1747ae917dd50ab20e6152 : SUCCESS in 1h 31m 24s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/24a5837ea46543c3b08ddfec1e82b04f : SUCCESS in 1h 41m 00s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ce75f8604fb42e68f021003ac8b0062 : SUCCESS in 1h 30m 38s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6dfd24dc33b4898bd77bcf287307690 : SUCCESS in 1h 38m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92fedd4e0497410b9fd2e38da326bf7c : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":1},{"id":"fd856657e6732c0489e7bd5481ffec2e72c535ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 11:45:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5130f787e66ab33c911828de292ebf0e725ef2d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-22 11:46:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb8cac5e1d347b29176fcf1f74da442\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbcda7ef9b3a4f89a87a0af82c2e2ab0 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea6f6b4fd3079af523770796c402306ba42f7ed5","revisions":{"ea6f6b4fd3079af523770796c402306ba42f7ed5":{"kind":"REWORK","_number":1,"created":"2026-04-20 12:03:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/985385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_heat","ref":"refs/changes/85/985385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/85/985385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/85/985385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/85/985385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_heat refs/changes/85/985385/1"}}},"commit":{"parents":[{"commit":"2a98cee8066b78c7ba66ca205ceb9fd232be42af","subject":"Pin setuptools via global-requirement-pins.txt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/2a98cee8066b78c7ba66ca205ceb9fd232be42af"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-04-19 15:38:33.000000000","tz":540},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-04-20 12:03:51.000000000","tz":0},"subject":"Add keystoneauth options to [ec2authtoken]","message":"Add keystoneauth options to [ec2authtoken]\n\n... otherwise the ec2tokens API fails due to the new policy rule to\nprohibit access without authentication.\n\nDepends-on: https://review.opendev.org/966092\nChange-Id: Ib36e9adf6f27fba662fed2dd955f586bde7b4c2d\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n(cherry picked from commit c08136ef5c10395a57477a99e1225f960f59d9db)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/ea6f6b4fd3079af523770796c402306ba42f7ed5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_heat/commit/ea6f6b4fd3079af523770796c402306ba42f7ed5"}]},"branch":"refs/heads/stable/2025.1","conflicts":{"ours":"2a98cee8066b78c7ba66ca205ceb9fd232be42af","theirs":"73d19b637aa8c92104a7807d702cdd571f8b2baf","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
