)]}'
{"id":"openstack%2Fopenstack-ansible-os_horizon~607474","triplet_id":"openstack%2Fopenstack-ansible-os_horizon~master~I9f1b8e4c6f5b4fe2601e5ee1dacf90bded60f93d","project":"openstack/openstack-ansible-os_horizon","branch":"master","topic":"bug/1795788","hashtags":[],"change_id":"I9f1b8e4c6f5b4fe2601e5ee1dacf90bded60f93d","subject":"Horizon configuration to support domains dropdown","status":"MERGED","created":"2018-10-03 06:40:00.000000000","updated":"2020-04-14 12:36:38.000000000","submitted":"2020-04-10 22:25:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"607474-1586557518150-1d1e633c","meta_rev_id":"6abb17a38f6de0e76cc3e612c9e6259d9d8fce7a","_number":607474,"virtual_id_number":607474,"owner":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-10 22:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-04-09 12:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},{"value":2,"date":"2020-04-07 14:36:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":1,"date":"2020-04-09 12:38:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 18:16:32.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-03-16 20:06:02.000000000","updated_by":{"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},"reviewer":{"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},"state":"REVIEWER"},{"updated":"2020-04-07 14:36:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-09 12:38:11.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-04-10 22:25:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5cadca3d50ad7b658983a2fe2bbd9531cebbaa9e","author":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"date":"2018-10-03 06:40:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fed17bb4ee7f0511efe3f636b9ef87c016626a57","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2018-10-03 06:40:11.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\n    Your patch will now be tested automatically by OpenStack testing frameworks\n    and once the automatic tests pass, it will be reviewed by other friendly\n    developers. They will give you feedback and may require you to refine it.\n\n    People seldom get their patch approved on the first try, so don\u0027t be\n    concerned if requested to make corrections. Feel free to modify your patch\n    and resubmit a new change-set.\n\n    Patches usually take 3 to 7 days to be reviewed so be patient and be\n    available on IRC to ask and answer questions about your work. Also it\n    takes generally at least a couple of weeks for cores to get around to\n    reviewing code. The more you participate in the community the more\n    rewarding it is for you. You may also notice that the more you get to know\n    people and get to be known, the faster your patches will be reviewed and\n    eventually approved. Get to know others and become known by doing code\n    reviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\n    Thanks again for supporting OpenStack, we look forward to working with you.\n\n    IRC: https://wiki.openstack.org/wiki/IRC\n    Workflow: http://docs.openstack.org/infra/manual/developers.html\n    Commit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"cdf88599e543b139e982b8909f1093def6dffc8d","author":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"date":"2018-10-03 06:44:10.000000000","message":"Topic changed from bug/1795788 to bug/1795788 Horizon configuration to support domains dropdown","accounts_in_message":[],"_revision_number":1},{"id":"7a9223d7726b8afaecdb1ccb6e321c04ab5423bf","author":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"date":"2018-10-03 06:44:39.000000000","message":"Topic changed from bug/1795788 Horizon configuration to support domains dropdown to bug/1795788","accounts_in_message":[],"_revision_number":1},{"id":"105f6abfd63ebbcd65fbec6177b196accfa60d44","author":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"date":"2018-10-03 06:49:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b8f8353b69eb01ac742886ed7f122f9733b32b75","author":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"date":"2018-10-03 06:52:27.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"517774b9382b03f284e206276ae1da81eeec46df","author":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"date":"2018-10-03 07:13:23.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c659c7df553b4ac45328b2ced84d4333ce3f83b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-03 07:57:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/74/607474/4/check/build-openstack-sphinx-docs/41cc3a0/html/ : SUCCESS in 3m 00s\n- openstack-ansible-linters http://logs.openstack.org/74/607474/4/check/openstack-ansible-linters/c2c2c83/ : SUCCESS in 6m 34s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-centos-7/1906b87/ : SUCCESS in 38m 29s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-opensuse-150/98f8cb4/ : FAILURE in 9m 14s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-opensuse-423/e38b0ba/ : FAILURE in 15m 06s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-ubuntu-bionic/594431b/ : SUCCESS in 29m 52s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-ubuntu-xenial/3b9b045/ : SUCCESS in 27m 36s\n- openstack-tox-docs http://logs.openstack.org/74/607474/4/check/openstack-tox-docs/eaad676/html/ : SUCCESS in 3m 45s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-distro_install-centos-7/e1ecf90/ : SUCCESS in 35m 11s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-distro_install-opensuse-423/677f17a/ : FAILURE in 17m 13s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-distro_install-ubuntu-xenial/461998a/ : SUCCESS in 30m 34s","accounts_in_message":[],"_revision_number":4},{"id":"6cf7132c160fc0fa5205b49e92ca01f7e1332d58","author":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"date":"2018-10-03 08:12:30.000000000","message":"Patch Set 4:\n\n\u003e Build failed (check pipeline).  For information on how to proceed,\n \u003e see\n \u003e http://docs.openstack.org/infra/manual/developers.html#automated-testing\n \u003e \n \u003e \n \u003e - build-openstack-sphinx-docs http://logs.openstack.org/74/607474/4/check/build-openstack-sphinx-docs/41cc3a0/html/\n \u003e : SUCCESS in 3m 00s\n \u003e - openstack-ansible-linters http://logs.openstack.org/74/607474/4/check/openstack-ansible-linters/c2c2c83/\n \u003e : SUCCESS in 6m 34s\n \u003e - openstack-ansible-functional-centos-7 http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-centos-7/1906b87/\n \u003e : SUCCESS in 38m 29s\n \u003e - openstack-ansible-functional-opensuse-150 http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-opensuse-150/98f8cb4/\n \u003e : FAILURE in 9m 14s (non-voting)\n \u003e - openstack-ansible-functional-opensuse-423 http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-opensuse-423/e38b0ba/\n \u003e : FAILURE in 15m 06s\n \u003e - openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-ubuntu-bionic/594431b/\n \u003e : SUCCESS in 29m 52s\n \u003e - openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-ubuntu-xenial/3b9b045/\n \u003e : SUCCESS in 27m 36s\n \u003e - openstack-tox-docs http://logs.openstack.org/74/607474/4/check/openstack-tox-docs/eaad676/html/\n \u003e : SUCCESS in 3m 45s\n \u003e - openstack-ansible-functional-distro_install-centos-7\n \u003e http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-distro_install-centos-7/e1ecf90/\n \u003e : SUCCESS in 35m 11s\n \u003e - openstack-ansible-functional-distro_install-opensuse-423\n \u003e http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-distro_install-opensuse-423/677f17a/\n \u003e : FAILURE in 17m 13s\n \u003e - openstack-ansible-functional-distro_install-ubuntu-xenial\n \u003e http://logs.openstack.org/74/607474/4/check/openstack-ansible-functional-distro_install-ubuntu-xenial/461998a/\n \u003e : SUCCESS in 30m 34s\n\nI seems that opensuse pipelines fails on \"[galera_server : Install percona repo]\" tasks.","accounts_in_message":[],"_revision_number":4},{"id":"d8f26f943f004ad9c195fca7cd91bde28f25bd92","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-03-23 15:57:58.000000000","message":"Patch Set 4:\n\nJust was looking through bugs and realised, that I accidentally committed the similar patch over there https://review.openstack.org/#/c/622256/2\n\nSo I\u0027d say, that it was partially resolved, except specifying list of available domains in dropdown. But if it\u0027s still relevant, you may rebase on the top of master in order to proceed with your part of the patch","accounts_in_message":[],"_revision_number":4},{"id":"8093abbcc07bb1e6e1a513c27eec85fc040d3bbc","author":{"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},"date":"2020-03-13 14:06:33.000000000","message":"Patch Set 4:\n\nIs there a way to get this small change merged for the next release?","accounts_in_message":[],"_revision_number":4},{"id":"f3d9e8957b2b2eeb5550355e03dd06276eecc6ac","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-13 17:59:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"b72d8dfcb152658ac8b563669a731a90908efc2c","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-03-13 18:05:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"241e821d868291cd3ca5d3dbed16cd8ff5721222","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-03-13 18:16:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7912c155f09bd7dd304b6a6f4c3333557b22a920","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 20:32:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/9b118083c441477ba17bc413756e3780 : SUCCESS in 2h 15m 42s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/422775e159cf48e38a1f76fc77db7e02 : SUCCESS in 2h 12m 28s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/642cd53b7bbf4b1a89d443bf20a2f2cc : SUCCESS in 2h 26m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/516a259a300443d594479bc864a0a3e5 : SUCCESS in 1h 13m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5abdfca343b745bb91c404d92ca9b2bf : SUCCESS in 1h 57m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f00a4b5f9d6d49c0a63b43314cbd6df9 : SUCCESS in 1h 39m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1f2aad523ef454289622d79646c4333 : SUCCESS in 1h 41m 25s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/eb44920d005d43f09d4a4df7f024ae65 : FAILURE in 8m 05s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df95579414f746148aaba8b66094d66f : SUCCESS in 2h 28m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9f7a58d1683849d3b888de86509f7167 : SUCCESS in 1h 10m 59s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/aa6cefe73c794181bb4bd01554545538 : FAILURE in 23m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84689ac2dc9e4d64b4aaedfbe8df15b6 : FAILURE in 50m 32s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fcb11744d64844b998b03c6744e361e4 : SUCCESS in 11m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623e12bc3adc4964b556053085147b68 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":5},{"id":"86eb250ca520f0ada71091a610ee8ce03d47fe8b","author":{"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},"date":"2020-03-16 13:28:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bb1bc8ddef3e1b14c4529260c6145162d918be75","author":{"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},"date":"2020-03-16 13:30:05.000000000","message":"Patch Set 6:\n\nI took the liberty of creating an updated version.","accounts_in_message":[],"_revision_number":6},{"id":"c4e134d4dcab9b2e1b5a328b84696677e0f55795","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-16 14:08:50.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)\n\nEdvin, thanks for taking care of it. I like the idea of extending functionality, just have some concerns about current implementation, since this change will change default behaviour, once horizon_keystone_multidomain_support is already true - so ppl would have dropdown only with one option until OPENSTACK_KEYSTONE_DOMAIN_CHOICES won\u0027t be defined. And omiting this variable will work just fine for many ppl.","accounts_in_message":[],"_revision_number":6},{"id":"ace6c2231cbc17df98bdfb08cd130add82453e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-16 16:32:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/54c33a3bb6b24c72a2aad4a4cbccc75f : SUCCESS in 2h 57m 19s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/4251d24e847740ef99a45527421a5a87 : SUCCESS in 1h 45m 40s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d749f9a7ba824acf8cc797cf5870fab3 : SUCCESS in 1h 54m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/381af566f47845e9a8d14fa90aa25510 : SUCCESS in 1h 04m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae3efea8649f49a6aaced3143bb7f67d : SUCCESS in 1h 30m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/25b154bcfdb645f2a2f0c0f647d1ca94 : SUCCESS in 1h 37m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1739eba3b11f4f128fcdf3d702236089 : SUCCESS in 1h 36m 33s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0295efdd883d4bdd8c2133ba3df00611 : FAILURE in 10m 40s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/63da15ce3d994b32bf16f5958fc04d74 : SUCCESS in 2h 19m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/048888aa33d0482ba1ab2773b766cd16 : SUCCESS in 1h 22m 46s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/853f173a375c4d1a912c017ce8e9fab2 : FAILURE in 45m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02529526ed2e44b58f74baa9526c85d6 : FAILURE in 43m 42s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/18332018b2b745c88994805d986a55dd : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea78dc9b291b4f208de8d703ce332310 : SUCCESS in 7m 51s","accounts_in_message":[],"_revision_number":6},{"id":"4e1d2466fb64ac44979c082db7aaba993886a296","author":{"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},"date":"2020-03-16 20:06:02.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nWhen I try the current functionality with horizon_keystone_multidomain_dropdown \u003d True I don\u0027t even get \"default\" in the dropdown. It is completely empty, so I cannot login, see my comment in the file.","accounts_in_message":[],"_revision_number":6},{"id":"e261a7c8eeb6f3733c37c576f6d4a42cab56af2a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-03 13:45:35.000000000","message":"Patch Set 6:\n\n(2 comments)\n\nSorry for the late reply\n\nYour comment looks pretty fair. Have you tried also to place patch for the horizon?\nUntil it\u0027s not fixed there I\u0027m pretty ok to do it that way, except small change for defaults","accounts_in_message":[],"_revision_number":6},{"id":"bad2a1f24962faf42eec981f93dbb70798e86b10","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-03 13:46:48.000000000","message":"Patch Set 6:\n\nBtw replying to your question about backporting - I think we can and should backport this to train.","accounts_in_message":[],"_revision_number":6},{"id":"2b4e2b4d555b01a05106182ee62c1ddb29a783e3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-07 14:35:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9c4643f3c40aed9c4f027a52a2153aa1e088d572","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-07 14:36:26.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"df69c861de5de3db120d01b6ea1c973d7d46d80f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 17:42:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/234794e85be74a33b3df04b0d6cd585b : SUCCESS in 2h 44m 28s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/e4d2d1120b754595a7e59b34b5f689cf : SUCCESS in 2h 00m 53s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f2219c3e88a4d219d92c837ab239178 : SUCCESS in 2h 18m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2f6f9ca4b0bf469b99cff6e344c849ca : FAILURE in 1h 10m 40s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b1c449a2df724104b3d90bf3157f5ce5 : SUCCESS in 1h 50m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a36d1fb709a244bfbddba74bdcf0ad68 : SUCCESS in 1h 42m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59ba5335489645268bbbe410e60a0c94 : SUCCESS in 1h 16m 55s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b5b2000a5a634ddd87f8f05077e5044e : SUCCESS in 2h 55m 49s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/107bd8c37bea40d09d0b1a359613e04c : SUCCESS in 2h 19m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d01dba51107947a682d2f501c7f1b413 : SUCCESS in 1h 16m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/da3df3bd3f804d068420501e3dd4d8a5 : FAILURE in 42m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8acefe33c16c46bfbf5dad3aadbaaf3e : SUCCESS in 1h 07m 31s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5214ad4faf5a4ad9946d00ce64ee9b55 : SUCCESS in 8m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f0ec4086c84ff2b80407367ac8adca : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":7},{"id":"ac89fbb55b64bb0a283f997ccd0b419f77fe2119","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-08 10:41:11.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"54e83008f28e8aaae3c79bb85b132587306cdd7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 13:36:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/a4d7c7a876104962bbe8625c1b7fe88d : SUCCESS in 2h 32m 19s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/1d146dc754d64bb58a082088f1eeaa27 : SUCCESS in 1h 54m 51s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22bc92fa5b0a4c42a8ba66d013d87b9e : SUCCESS in 2h 01m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/46e36fd1533c4f9eb7aa25bfe51825bb : SUCCESS in 1h 30m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7c3e7ac54a0e4376898938fb4c4a92f6 : SUCCESS in 1h 34m 53s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/73378345d15a4ac9b9e6b2ae1b028bfe : SUCCESS in 1h 24m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7014747c3fa94314aacbb3e09e22b7c2 : SUCCESS in 1h 19m 50s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f19d3949cc2d46dea57efe513d69dade : SUCCESS in 2h 46m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2427caf075514c71ad46b1fd4314c4e2 : SUCCESS in 2h 11m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a327db40f776425f9713c956c81af91b : SUCCESS in 1h 22m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2ecf880458094f75a194aa9ffb2a2a52 : FAILURE in 43m 48s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ee3feee0d4b4448b60ed11441e7a700 : SUCCESS in 1h 36m 56s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e3d5913751424d9fa8d6a16aa1cbf327 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f64a6159f78477c80ae009ed23a3a43 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":7},{"id":"c55f6b0b7cd3b8f404d5f2e17d5fc9737fdadec2","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-04-09 12:38:11.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"d1a0c84aff7a9541fb823554781e6e5a9f051f5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 12:38:23.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2e6ed62bb6da0d3b79fcbf45b07faeae294734a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 15:20:34.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/2f018a85538a40a891f879c8649f6aa3 : SUCCESS in 2h 35m 47s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c30fc73543b24ec99d6467ecaf7610df : SUCCESS in 1h 51m 46s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/908c1770b39a40f5afec6d55364bcbda : SUCCESS in 2h 10m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c09ab8491ad445b18a3b395ad7752324 : SUCCESS in 1h 16m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ee8d97ba260b4e30b36eb89edd41ee4d : SUCCESS in 1h 39m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b5af125ec56a43a297a0398609a62c1c : FAILURE in 1h 25m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0434f18730944a0389b4075bed4f48f1 : SUCCESS in 1h 25m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0418db16153d4f1fb13f5b742c6614e8 : SUCCESS in 1h 15m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17c70f1eb38740368fe0f72f39dd45bb : SUCCESS in 1h 07m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0126cc543cf4634aea7459774ea9327 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":7},{"id":"9e2d84b3c04013468d5526913b19eae2bdb9333b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-09 18:18:20.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"b6fc35650549b43b76604aad69bea0ae5d08a2ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 21:13:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/16b7214e36ec4f80b40b39cfa609b612 : SUCCESS in 2h 47m 01s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/18a13ae451f74615be6dbcf40a7119cc : SUCCESS in 1h 57m 11s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa1d31a4ea1c45e380dcb0a3b9f6c708 : SUCCESS in 1h 54m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7a92243f38d24dfda6b1c5b285744aa3 : SUCCESS in 1h 09m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/92a928095c25463288d2d3bedf4228be : POST_FAILURE in 1h 17m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d6c218d9330d43608b125fdeefb550c6 : SUCCESS in 1h 32m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98fbd98f4cfc42b3a4041ac417d4e8d3 : SUCCESS in 1h 23m 22s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a0d1cad75ea446db95200e23b5afac66 : SUCCESS in 2h 44m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd12e23a2f5a47feb7428d3fde836618 : SUCCESS in 2h 09m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a108cddf747c4c2f9fcb118a9d830198 : SUCCESS in 1h 24m 09s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0fb3a6b7161a469caaa3b00cf47cdae7 : FAILURE in 45m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5d4edcf5d6c4b6eb406f9b4ad649543 : SUCCESS in 1h 00m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31605d8b24064e1c8a62097366256fa0 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":7},{"id":"00ebc24ccad435c231644684925aef4a835c1e81","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-10 07:01:25.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"810e6e1a387890f13d6f6d13a7fe40ba3601d8de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 13:11:32.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/c0b7b642958441cfb17c4624a6b4c76a : SUCCESS in 2h 43m 51s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/f08d3229c9eb4a40bd090a0ca541e328 : SUCCESS in 1h 44m 25s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae89189dfdb448e1ba6af7a0ad1e0e91 : SUCCESS in 1h 47m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2350aba99a424741a2b288134780febe : SUCCESS in 1h 14m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/42d0e101207d41b08e95ab8dda708e9c : SUCCESS in 1h 46m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b92a166f25fa4e0da2f996f60ed1867d : FAILURE in 1h 26m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daa0a0b13f8844c9ae8c5779ef7da94d : SUCCESS in 1h 46m 36s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c14b93d0b67149b5b5ad35d1438fc781 : SUCCESS in 2h 41m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2210e0dab8d64fc8a83d90f8b91b8f70 : SUCCESS in 2h 14m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/25dba077cb644a62b46ef350703c6765 : SUCCESS in 1h 31m 12s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/31dc9d227bb048559dec24c09ad80883 : FAILURE in 23m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ccc62279877456bb4e3f010ec598b51 : SUCCESS in 1h 31m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83199815a7fe48168130b25dbc817cc6 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":7},{"id":"8b0c6f4e3b389ae641ecd0dbc536a34e4302487d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-10 17:02:42.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"0e5b8998d6ce0358bcf655315478db2e80043df0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 19:49:46.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/f5d0d31c41a14940a0e3e1fa28372171 : SUCCESS in 2h 17m 10s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6d109562e3a34d56b32c41efa468478f : SUCCESS in 1h 44m 19s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/143b9f94fe3d4a0194724558c53ba3f3 : SUCCESS in 1h 56m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8454fca35ca24378925d2ab812160735 : SUCCESS in 1h 07m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0bd825dd3e544245853b9bb8f74aa525 : SUCCESS in 1h 33m 46s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9e3b88a19ee648c2a9792ea4f3a6028e : SUCCESS in 1h 31m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbeaf529090a412d9bca017401425d2a : SUCCESS in 1h 17m 05s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/284343053995459398d08bce6bfa1bc1 : SUCCESS in 2h 45m 26s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94f4a1482d4c4c20ae8a018302aee923 : SUCCESS in 1h 59m 33s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a47e7358f63d45d2bc23c6d0e3f6fe2e : SUCCESS in 1h 08m 31s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ff2965ab9b19444b8599de97013c5690 : FAILURE in 35m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a038a32be444e5386e9c5877b506e2b : SUCCESS in 58m 35s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9fd46c73a0cf4087be54a90fee730914 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ec3865747c34e1baca10a8518b86c6b : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":7},{"id":"7da2023bdb681f06181fd399924b074a974403f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 19:49:57.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8277f5c5fc8ea86b29480bd1b878cf3fd94b55e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 22:25:17.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/d196e124a0f04a038970d5f7bd265c49 : SUCCESS in 2h 33m 28s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/fb7514ca775d4932a8d38c10eea5ce91 : SUCCESS in 2h 00m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eac31124782c4b43825e61f1a4527864 : SUCCESS in 1h 55m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/93594153e22d44a19ec12dce3664364c : SUCCESS in 1h 06m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4b4f99484d594d95b3e68d0e47ff4fa2 : SUCCESS in 1h 32m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fe7efa43d51b4626941a5832ea83aecc : SUCCESS in 1h 22m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/561bfef115dd491e98318dc4f4cc8a4f : SUCCESS in 1h 14m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2346bb5ed16349f8bac81cbae3d7409a : SUCCESS in 1h 10m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17882e516f9f41ff9d747f88b57542dd : SUCCESS in 58m 35s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9342a2b696a14a15af059b3dd5446b16 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e7812380874dab851957b56b463e44 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":7},{"id":"b078f1134b0223a1ba7cf69e6c356edd17ffdaeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 22:25:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"979cdfc61b22bc10108283821ccab96c78c3a3e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 22:26:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df075c70cf0c402c9f763dc48758d10e : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":7},{"id":"712d08bf9be5a3e7a3d1d5f0eb18c818d153ffdb","author":{"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},"date":"2020-04-14 07:46:08.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nExcellent, thanks! How would the process be for backporting to train?","accounts_in_message":[],"_revision_number":6},{"id":"6abb17a38f6de0e76cc3e612c9e6259d9d8fce7a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-14 12:36:38.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 5c7b59699548575f6a76a421bebc979965058704","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"d86285b377221fb5522207a9da6cbfbdea4dd1e9","revisions":{"22b140be2bd55d911b8184ff476a4157756a44eb":{"kind":"REWORK","_number":1,"created":"2018-10-03 06:40:00.000000000","uploader":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"ref":"refs/changes/74/607474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/74/607474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/1"}}},"commit":{"parents":[{"commit":"d1749ae04f8ff25086d6dccbc9b2dbb5c7e4bdaf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/d1749ae04f8ff25086d6dccbc9b2dbb5c7e4bdaf"}]}],"author":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:40:00.000000000","tz":0},"committer":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:40:00.000000000","tz":0},"subject":"An openstack-ansible deployment does not have any options to set the local_settings.py options for Horizon to support Keystone V3 multiple domains dropdown menu and its variants to choose. These options in local_settings.py are:","message":"An openstack-ansible deployment does not have any options to set the local_settings.py options for Horizon to support Keystone V3 multiple domains dropdown menu and its variants to choose.\nThese options in local_settings.py are:\n\nOPENSTACK_KEYSTONE_DOMAIN_DROPDOWN - default value to \"False\"\nOPENSTACK_KEYSTONE_DOMAIN_CHOICES - default value to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nThe proposed solution is to add two variables in the horizon role (user_variables.yml):\n\nhorizon_keystone_domain_dropdown - defaults to \"False\"\nhorizon_keystone_domain_choices - defaults to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nChange-Id: I9f1b8e4c6f5b4fe2601e5ee1dacf90bded60f93d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/22b140be2bd55d911b8184ff476a4157756a44eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/22b140be2bd55d911b8184ff476a4157756a44eb"}]},"branch":"refs/heads/master"},"899a00fb51700e792b75e186687e0189a1ca088b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-10-03 06:49:31.000000000","uploader":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"ref":"refs/changes/74/607474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/74/607474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/2"}}},"commit":{"parents":[{"commit":"d1749ae04f8ff25086d6dccbc9b2dbb5c7e4bdaf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/d1749ae04f8ff25086d6dccbc9b2dbb5c7e4bdaf"}]}],"author":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:40:00.000000000","tz":0},"committer":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:45:29.000000000","tz":0},"subject":"Horizon configuration to support domains dropdown","message":"Horizon configuration to support domains dropdown\n\nAn openstack-ansible deployment does not have any options to set the local_settings.py options for Horizon to support Keystone V3 multiple domains dropdown menu and its variants to choose.\nThese options in local_settings.py are:\n\nOPENSTACK_KEYSTONE_DOMAIN_DROPDOWN - default value to \"False\"\nOPENSTACK_KEYSTONE_DOMAIN_CHOICES - default value to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nThe proposed solution is to add two variables in the horizon role (user_variables.yml):\n\nhorizon_keystone_domain_dropdown - defaults to \"False\"\nhorizon_keystone_domain_choices - defaults to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nChange-Id: I9f1b8e4c6f5b4fe2601e5ee1dacf90bded60f93d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/899a00fb51700e792b75e186687e0189a1ca088b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/899a00fb51700e792b75e186687e0189a1ca088b"}]},"branch":"refs/heads/master"},"fcc6972e26465bf942bcaa192003c2a14e13df36":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2018-10-03 06:52:27.000000000","uploader":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"ref":"refs/changes/74/607474/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/74/607474/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/3"}}},"commit":{"parents":[{"commit":"d1749ae04f8ff25086d6dccbc9b2dbb5c7e4bdaf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/d1749ae04f8ff25086d6dccbc9b2dbb5c7e4bdaf"}]}],"author":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:40:00.000000000","tz":0},"committer":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:52:19.000000000","tz":0},"subject":"Horizon configuration to support domains dropdown","message":"Horizon configuration to support domains dropdown\n\nAn openstack-ansible deployment does not have any options to set the local_settings.py options for Horizon to support Keystone V3 multiple domains dropdown menu and its variants to choose.\nThese options in local_settings.py are:\n\nOPENSTACK_KEYSTONE_DOMAIN_DROPDOWN - default value to \"False\"\nOPENSTACK_KEYSTONE_DOMAIN_CHOICES - default value to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nThe proposed solution is to add two variables in the horizon role (user_variables.yml):\n\nhorizon_keystone_domain_dropdown - defaults to \"False\"\nhorizon_keystone_domain_choices - defaults to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nChange-Id: I9f1b8e4c6f5b4fe2601e5ee1dacf90bded60f93d\nCloses-Bug: 1795788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/fcc6972e26465bf942bcaa192003c2a14e13df36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/fcc6972e26465bf942bcaa192003c2a14e13df36"}]},"branch":"refs/heads/master"},"5da513ab2faa6ce9ca37ca439027914c7933771b":{"kind":"REWORK","_number":4,"created":"2018-10-03 07:13:23.000000000","uploader":{"_account_id":29230,"name":"Max Kochubey","email":"maxkochubey@gmail.com"},"ref":"refs/changes/74/607474/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/74/607474/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/4"}}},"commit":{"parents":[{"commit":"d1749ae04f8ff25086d6dccbc9b2dbb5c7e4bdaf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/d1749ae04f8ff25086d6dccbc9b2dbb5c7e4bdaf"}]}],"author":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:40:00.000000000","tz":0},"committer":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 07:12:52.000000000","tz":0},"subject":"Horizon configuration to support domains dropdown","message":"Horizon configuration to support domains dropdown\n\nAn openstack-ansible deployment does not have any options to set the local_settings.py options for Horizon to support Keystone V3 multiple domains dropdown menu and its variants to choose.\nThese options in local_settings.py are:\n\nOPENSTACK_KEYSTONE_DOMAIN_DROPDOWN - default value to \"False\"\nOPENSTACK_KEYSTONE_DOMAIN_CHOICES - default value to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nThe proposed solution is to add two variables in the horizon role (user_variables.yml):\n\nhorizon_keystone_domain_dropdown - defaults to \"False\"\nhorizon_keystone_domain_choices - defaults to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nChange-Id: I9f1b8e4c6f5b4fe2601e5ee1dacf90bded60f93d\nCloses-Bug: 1795788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/5da513ab2faa6ce9ca37ca439027914c7933771b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/5da513ab2faa6ce9ca37ca439027914c7933771b"}]},"branch":"refs/heads/master"},"0184594e95b1d14ff705a6f2d2c097d64db2c02e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-03-13 17:59:42.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/74/607474/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/74/607474/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/5"}}},"commit":{"parents":[{"commit":"cfb6365b75194cef94324e68bc9c506878e37b80","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/cfb6365b75194cef94324e68bc9c506878e37b80"}]}],"author":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:40:00.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-13 17:59:42.000000000","tz":0},"subject":"Horizon configuration to support domains dropdown","message":"Horizon configuration to support domains dropdown\n\nAn openstack-ansible deployment does not have any options to set the local_settings.py options for Horizon to support Keystone V3 multiple domains dropdown menu and its variants to choose.\nThese options in local_settings.py are:\n\nOPENSTACK_KEYSTONE_DOMAIN_DROPDOWN - default value to \"False\"\nOPENSTACK_KEYSTONE_DOMAIN_CHOICES - default value to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nThe proposed solution is to add two variables in the horizon role (user_variables.yml):\n\nhorizon_keystone_domain_dropdown - defaults to \"False\"\nhorizon_keystone_domain_choices - defaults to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nChange-Id: I9f1b8e4c6f5b4fe2601e5ee1dacf90bded60f93d\nCloses-Bug: 1795788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/0184594e95b1d14ff705a6f2d2c097d64db2c02e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/0184594e95b1d14ff705a6f2d2c097d64db2c02e"}]},"branch":"refs/heads/master"},"b56a99126b0cdda947d3f41c519d7ec018c4b3b9":{"kind":"REWORK","_number":6,"created":"2020-03-16 13:28:47.000000000","uploader":{"_account_id":31722,"name":"Edvin Linge","email":"edvin.linge@gmail.com","username":"elinge"},"ref":"refs/changes/74/607474/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/74/607474/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/6"}}},"commit":{"parents":[{"commit":"cfb6365b75194cef94324e68bc9c506878e37b80","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/cfb6365b75194cef94324e68bc9c506878e37b80"}]}],"author":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:40:00.000000000","tz":0},"committer":{"name":"Edvin Linge","email":"edvin.linge@gmail.com","date":"2020-03-16 13:24:17.000000000","tz":60},"subject":"Horizon configuration to support domains dropdown","message":"Horizon configuration to support domains dropdown\n\nAn openstack-ansible deployment does not have any options to set the local_settings.py options for Horizon to support Keystone V3 multiple domains dropdown menu and its variants to choose.\nThese options in local_settings.py are:\n\nOPENSTACK_KEYSTONE_DOMAIN_DROPDOWN - default value to \"False\"\nOPENSTACK_KEYSTONE_DOMAIN_CHOICES - default value to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nThe proposed solution is to add two variables in the horizon role (user_variables.yml):\n\nhorizon_keystone_multidomain_dropdown - defaults to \"False\"\nhorizon_keystone_multidomain_choices - defaults to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nChange-Id: I9f1b8e4c6f5b4fe2601e5ee1dacf90bded60f93d\nCloses-Bug: 1795788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/b56a99126b0cdda947d3f41c519d7ec018c4b3b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/b56a99126b0cdda947d3f41c519d7ec018c4b3b9"}]},"branch":"refs/heads/master"},"d86285b377221fb5522207a9da6cbfbdea4dd1e9":{"kind":"REWORK","_number":7,"created":"2020-04-07 14:35:41.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/74/607474/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/74/607474/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/74/607474/7"}}},"commit":{"parents":[{"commit":"cfb6365b75194cef94324e68bc9c506878e37b80","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/cfb6365b75194cef94324e68bc9c506878e37b80"}]}],"author":{"name":"Max Kochubey","email":"maxkochubey@gmail.com","date":"2018-10-03 06:40:00.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-04-07 14:24:11.000000000","tz":180},"subject":"Horizon configuration to support domains dropdown","message":"Horizon configuration to support domains dropdown\n\nAn openstack-ansible deployment does not have options to set the local_settings.py params\nfor Horizon to support Keystone V3 multiple domains dropdown menu and its variants to choose.\n\nThese options in local_settings.py are:\nOPENSTACK_KEYSTONE_DOMAIN_CHOICES - default value to \"((\u0027Default\u0027, \u0027Default\u0027))\"\n\nChange-Id: I9f1b8e4c6f5b4fe2601e5ee1dacf90bded60f93d\nCloses-Bug: 1795788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/d86285b377221fb5522207a9da6cbfbdea4dd1e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/d86285b377221fb5522207a9da6cbfbdea4dd1e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
