)]}'
{"id":"openstack%2Fopenstack-ansible-os_horizon~682018","triplet_id":"openstack%2Fopenstack-ansible-os_horizon~master~Ic39a4d1cc03df72a4080abd6513bf8f4036da456","project":"openstack/openstack-ansible-os_horizon","branch":"master","topic":"retire_lbaas","hashtags":[],"change_id":"Ic39a4d1cc03df72a4080abd6513bf8f4036da456","subject":"Retire neutron-lbaas UI plugin","status":"MERGED","created":"2019-09-13 10:38:06.000000000","updated":"2019-09-20 16:50:54.000000000","submitted":"2019-09-20 16:49:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682018-1568998180499-b80b591b","meta_rev_id":"713a27c05e97b4252b797116425c6958a9f731e7","_number":682018,"virtual_id_number":682018,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-09-20 16:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-09-20 13:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 19:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-09-20 13:27:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 19:21:09.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-09-20 13:27:48.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-09-20 16:49:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"459bea72af773f404c84ffc5690ac4b1c8063f23","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-13 10:38:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2992fea73a295c4995b1a3d2351251307f5bed6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 13:52:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/13868e1b134443c5aa594e752308403e : SUCCESS in 3h 08m 08s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/d4e6fab7d38941018a1d7a1baf220244 : SUCCESS in 2h 12m 08s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1cba176ecdb94885ae806d05d414500b : SUCCESS in 1h 55m 55s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/92221f5c50e5435ebf6add77336e2a68 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d053ebb7c3844de90e85b688f4ce9e3 : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f3c4c41cd0a4297b4e2352457a6ed79 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b261bb16d0f3144425463dd93de8f0baa53784a4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-09-18 19:21:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"65bb91936d1c06923cb91752dd1e80e15a4e2be5","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-09-20 13:27:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a660c9bdc19e7a8800fe11a1f6d3aada629c3eac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:28:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c06e15c4bed831c00aede9524dea21d97fc9feea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:49:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-7 https://zuul.opendev.org/t/openstack/build/091df954eede425490ba1eeab32d447a : SUCCESS in 3h 15m 12s\n- openstack-ansible-deploy-aio_lxc-debian-stable https://zuul.opendev.org/t/openstack/build/522ae766b28d4a32b09b0ef81fc5c5fb : SUCCESS in 2h 09m 50s\n- openstack-ansible-deploy-aio_lxc-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3725997e53a4873813d69e2915e8492 : SUCCESS in 2h 18m 51s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4c4f9c71e66442e8a400ac8d0d077f6a : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9191f5ed1a3c4702a3453c3eeb256455 : SUCCESS in 3m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43782ac1ee18419d80e6dee340c4b1fd : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4e178b06e81f564a7e169cfcdebd94283e39dd32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:49:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6a96d351c435bba7cc9d9780ea6e62f55e3aa99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:50:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf93cd2ad9c4b3fa33e5d95e12e3bc9 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b5aaf7fd3324d6cac8723498c3eb8b2 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2280e2e903bdad3f6f4e4675c01acc4ccfcd0bf","revisions":{"a2280e2e903bdad3f6f4e4675c01acc4ccfcd0bf":{"kind":"REWORK","_number":1,"created":"2019-09-13 10:38:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/18/682018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/18/682018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/18/682018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/18/682018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/18/682018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/18/682018/1"}}},"commit":{"parents":[{"commit":"719cdbac4c969f18bf095773e3bdb0100b81c076","subject":"Merge \"Allow venv python interpreter to be overridden\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/719cdbac4c969f18bf095773e3bdb0100b81c076"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-13 10:37:54.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-09-13 10:37:54.000000000","tz":180},"subject":"Retire neutron-lbaas UI plugin","message":"Retire neutron-lbaas UI plugin\n\nBecause of global retirement process of neutron-lbaas  we stop providing\nneutron-lbaas-ui plugin installation for horizon.\n\nChange-Id: Ic39a4d1cc03df72a4080abd6513bf8f4036da456\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/a2280e2e903bdad3f6f4e4675c01acc4ccfcd0bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/a2280e2e903bdad3f6f4e4675c01acc4ccfcd0bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
