)]}'
{"id":"openstack%2Fopenstack-ansible-os_horizon~754169","triplet_id":"openstack%2Fopenstack-ansible-os_horizon~master~I2c60a5ed861b3d8f364668ef9efb858a7cc895f5","project":"openstack/openstack-ansible-os_horizon","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I2c60a5ed861b3d8f364668ef9efb858a7cc895f5","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-09-24 16:56:13.000000000","updated":"2020-10-12 10:20:00.000000000","submitted":"2020-10-12 10:18:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"754169-1602497911572-adfe8283","meta_rev_id":"71ba0e4f59d00c7d97f4bdb94dd3490624c8e334","_number":754169,"virtual_id_number":754169,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-12 10:18:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-12 05:36:58.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2020-10-05 07:30:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2020-09-25 16:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2020-10-05 07:30:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2020-10-09 09:41:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-05 07:30:21.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2020-10-09 09:41:35.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-10-12 05:36:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-10-12 10:18:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3400acf7478874dc8edfc48aff02c2e5147f5f4d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-09-24 16:56:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9d0888947a331327947e6c68c142645abd01322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 19:37:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/0f3dd6373fc24e1594409de37ce1524b : SUCCESS in 2h 03m 12s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/efd37095e29a45f6bb5c72d23194d851 : SUCCESS in 1h 39m 51s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/609e8e3076764879bfc675c3e51e6c1c : SUCCESS in 2h 03m 59s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/138d9f9354ea43af9aeb923e60fe1700 : SUCCESS in 1h 31m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e264e36128e54977999e5d1ad48355fd : SUCCESS in 1h 12m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/48cf1dfd8c974f3fab742624d294229d : SUCCESS in 1h 34m 01s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42ea17e7c7fc49f19366a85a985c697b : SUCCESS in 1h 13m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2ad5a4838bb4eb4a26c23a87783c468 : SUCCESS in 1h 07m 57s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d1a2659d09154c9e931cb5aca50efd51 : SUCCESS in 2h 03m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f99fe5788c94c1b935f23515d064e75 : FAILURE in 49m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/35bef327f7174046a25e18026cb89f9e : FAILURE in 44m 07s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e12ff55e1a64068b842ede1ada612be : SUCCESS in 54m 06s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c7c4072db8ef4730988bf4c2f7e27c54 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9bf55ee4a65439b88106c32e97c1c47 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7afbea7a0cf10286ae6b180f5c200daab447740c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-24 21:25:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2197d3644ad26c1e322e3c1dc60d1a799897b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 00:13:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/4ecff40ed2f547ad8a56393b75092817 : SUCCESS in 1h 55m 42s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/6f9d2803159d47c2bbda98040c74a759 : SUCCESS in 1h 51m 30s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20b4114f74f44068971f4ba9cb00f502 : SUCCESS in 1h 29m 22s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbdbf066abf245af92389e4c8df7bc88 : SUCCESS in 1h 28m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/baac3df249f448df97d30fd202ba1722 : SUCCESS in 1h 13m 09s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0f0128315d294c9e8799d4656213cea1 : SUCCESS in 1h 33m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a146fed6b1994623966316fd4015e0c5 : SUCCESS in 1h 14m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08bde46654a446c6ba6f1ad55193941b : SUCCESS in 1h 12m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9b32bd06b9ab4bf584d440faff18ccbe : SUCCESS in 2h 09m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/116b7074e57e4c97a636b2ea9c8feea7 : SUCCESS in 2h 00m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dd453a1a766a45f38b4c95fd6afbcc4f : FAILURE in 52m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a8764e3e7994b08b9933ccaeb8f9f9d : SUCCESS in 55m 18s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/62bc54df52f24d18bb6783677a8ae1a9 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b30dde8af6924e1a9c83ebd6e8f15f1e : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"565b21686de5c13029af424cd6529158614ae176","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-25 07:33:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c3a77762b10d1fd5e83fc9344b0b2ec4662ee67","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-25 07:33:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b0e951586dfadfe2b739895d8c87be4fca32d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 09:40:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/1f09bcdff24f4934846ea059e9043d56 : FAILURE in 29m 39s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/d21381f0c55a49e2af8f9e0157c85383 : SUCCESS in 1h 36m 55s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25862a3dc66a44f4a003ae304b00d45b : SUCCESS in 1h 35m 49s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a974689ff66347a09e1cd72b14e72cb7 : SUCCESS in 1h 31m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/471a1b89115f4f07ae5f00159a6d1616 : POST_FAILURE in 1h 37m 00s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3cf6f6182e9b49a59cd6343a4a9c9272 : SUCCESS in 1h 21m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce9c6df73bd24bacaf0444fbcda0bd0c : SUCCESS in 1h 21m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b9e471a2b54474cb06ea5c191d0404c : POST_FAILURE in 1h 35m 47s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6eb84472a3c34bbbb1847486cb19760b : SUCCESS in 1h 53m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e32cc00f5bd44ecc8f487bcede995329 : SUCCESS in 1h 44m 27s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ceaca94c38584a198ccc6cd221b83502 : FAILURE in 48m 30s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ced79479d664ee88f0e96d72968d44f : SUCCESS in 57m 15s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c3cce8a461064029ba67831e7ebc1e68 : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25de2dbd0f9c451eb84327a2e28b160f : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":1},{"id":"51f35c5e797eef47709e927dfac2a16d1df1fb03","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-25 12:57:21.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3ed8e997a1054d3e1d063b2918f08198536e769e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 15:18:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/27a3fefe363b48d3af4329a43f3f6038 : SUCCESS in 1h 54m 40s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/c3875f5828b74a40b81d6363c9091542 : SUCCESS in 1h 42m 20s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e13ce4a5a65d4c9580f986665eb3c564 : SUCCESS in 1h 32m 35s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36a3038f43e447b9af3fc33bc3daa2db : SUCCESS in 1h 26m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/91f5b24966974806a7a345848772d118 : SUCCESS in 1h 12m 44s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/32bc5e0007db4d3a8669ccc81b854ad6 : SUCCESS in 1h 34m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/780d9e8f48a947348273dace10b530a8 : SUCCESS in 1h 09m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01675fb0a9b34f75a4182cbaa703c21a : SUCCESS in 1h 06m 46s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fabc337ab83b46509bf76a3a2c145e3c : SUCCESS in 2h 09m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41d6c920e51e40fca29656ba3eb0acd0 : SUCCESS in 1h 41m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/26a1e8f97b6141df9cdad90116e6dddc : SUCCESS in 1h 26m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf5ef82b11a94722bde2db300a1aa61a : SUCCESS in 1h 01m 50s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b303a0dab547410e94fd4380d4ad0fab : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6106def6132d4162bce8fbbc71674212 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ab65bf360819ab3362d18a4e49b8be27d418e858","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-25 16:22:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e35fde5858ad3e0b80af373b4fb7c2def66af135","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2020-10-05 07:30:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d0f36e7afc67640834a6e1f083d0f30fca3196ca","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-10-09 09:41:35.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"662c206729fa49ddf3153ca5d85118d98c29ac13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 09:41:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6645666e350c789d81096b561c863e32b07537b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 11:15:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/aa0031dc0baa401fbdc3097033623c4c : FAILURE in 27m 51s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/a755842aa1774bfb988f82894b088686 : FAILURE in 28m 48s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66ec5cc0b76f42f285ee0c496844c21a : FAILURE in 24m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c5a2cf1728044adca1de0399a01f180a : SUCCESS in 1h 09m 19s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c2b712c5720e41c8903fe3814a9493f0 : SUCCESS in 1h 28m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe4763d1e12c4bbd98ed681574616bea : SUCCESS in 1h 11m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/904d669f18c54c4fb42a2a4c985c80ed : SUCCESS in 1h 15m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d617ee3e316f41d9b957eea187e17187 : SUCCESS in 1h 06m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f941751370f4d05950fc3ee65b9cad4 : SUCCESS in 56m 21s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dfcc0924ee0848ac94e0fd6e301aa70f : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0739638f730c4883869f628f54b336f4 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":2},{"id":"45576cecaeb4ec849eae574e05cafcc9e454a7a3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-09 19:51:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0eec667ef923a80e10b70fbbbcd9863cbb195546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 22:25:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/9c5b45a628644a25aecd4f576fe0a5a0 : FAILURE in 1h 51m 31s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/9ae518cdea524e58bcb69b7975f18951 : SUCCESS in 1h 50m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bebbfbd23de149a4bea0d5f69d95758d : RETRY_LIMIT in 11m 26s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7af20f90b1c4440993b9bc014692760 : RETRY_LIMIT in 10m 54s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b78d086c6be8430cbbc5a8311c5746cb : SUCCESS in 1h 15m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5dc5dee5e0924e70976407e652e223cf : SUCCESS in 1h 23m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0923a550e4cd47cd8a95a69cccfd94cb : SUCCESS in 1h 14m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c38b9ace8ffa4b3882cb2d0ae96d4bd1 : SUCCESS in 1h 08m 18s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4af822f554ab4cbc8f6b872286d1ea64 : SUCCESS in 2h 16m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c7f4d0b7a104053b916bedc7a3e38d8 : SUCCESS in 1h 41m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/26a3e2d3b62644f3a8c70797f41656c3 : SUCCESS in 1h 04m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55379ab8d14241fcb9d812a62ed2c466 : SUCCESS in 1h 17m 38s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/540b522cee104bd5b7ff1346b0983ef8 : FAILURE in 18m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf640177fda24c508a20d9512c205995 : RETRY_LIMIT in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"51c6dd28d85272088a46bc70124e55a2b11a469c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-10-12 05:36:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d8efd2d66a0f9f98b09dcaf52fd4c8984d0f7dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 07:51:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/6f283018818c42c58e999902598a92a1 : SUCCESS in 1h 58m 24s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/504d0e5bb4834267a2badd81af1b3144 : SUCCESS in 1h 39m 43s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0c351e4b2564a91b17e8de294efd72d : SUCCESS in 1h 41m 29s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e02d8ed5c8a74fc09f86a274561e3c7c : SUCCESS in 1h 36m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f497e916d6e549179779518e570dceaa : SUCCESS in 1h 09m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cc5d266647114fdd992a6b01aad396fe : SUCCESS in 1h 24m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/57c9758b28ee494a962528c2cd97b415 : SUCCESS in 1h 15m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d71d79cd49242d9b15a861c2b8d9695 : SUCCESS in 1h 07m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3e0b171c3e3d4b12adf8af30d5147ad9 : SUCCESS in 2h 07m 13s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/465e10f51aa94fe29e3d1d111cf3cf1e : SUCCESS in 1h 43m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/778bdae2980a4be0bb1ba4fe0f711327 : SUCCESS in 1h 04m 34s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f163cafc49b74bcd80998b672059b4ee : SUCCESS in 53m 57s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/13f26271f2d14186a168809c894b39f0 : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2625a4d38f450b803abcb62e283dc6 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a99379bf1fe790fed31993206aca4eeb8db05c55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 07:51:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81834176e97fbd0ad3f65e33d2af90174d752387","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 10:18:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c99f0c1eb020da511a3f0b6b0bb1fbd558b2265a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 10:18:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/493e198b609c441399762dda2d0148a8 : SUCCESS in 1h 58m 37s\n- openstack-ansible-deploy-aio_lxc-debian-buster https://zuul.opendev.org/t/openstack/build/86b0b901a6c041a8a4a18afd595a86cc : SUCCESS in 1h 33m 14s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a9c66eda1e844a5be94989addf443d9 : SUCCESS in 1h 37m 07s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d314de82b22a446da258adf14f91a0c4 : SUCCESS in 1h 08m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8d77e5b1089a40aa845868d82a4db61f : SUCCESS in 1h 22m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6eddabb83cbb450d9a899a0b898b069c : SUCCESS in 1h 26m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f96eae27f8984ece952816d2b942e2f7 : SUCCESS in 1h 08m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/18b5fe2cc7104a37be9e2edf42d203ad : SUCCESS in 1h 08m 03s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf6b1e86225e457695a62547c39a0912 : SUCCESS in 58m 41s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fd832950cd2a4b15858e08e4f7cb47c9 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769d155c288644daa586a851a4fbe5ea : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"bc8ec60caca8bfeb8335c49283e030d5f54b2135","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-12 10:20:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/088e55ff53244852b26ce902ddee9503 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"be35f492ff030b6f03425639891a09bcf2619743","revisions":{"4cbcf87e1db6833a1858e8a4f08e509b0c52606e":{"kind":"REWORK","_number":1,"created":"2020-09-24 16:56:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/754169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/69/754169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/69/754169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/69/754169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/69/754169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/69/754169/1"}}},"commit":{"parents":[{"commit":"4c2cc73d571dd86303bc16063951c13bec8af8c5","subject":"Add Centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/4c2cc73d571dd86303bc16063951c13bec8af8c5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 16:56:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 16:56:11.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2c60a5ed861b3d8f364668ef9efb858a7cc895f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/4cbcf87e1db6833a1858e8a4f08e509b0c52606e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/4cbcf87e1db6833a1858e8a4f08e509b0c52606e"}]},"branch":"refs/heads/master"},"be35f492ff030b6f03425639891a09bcf2619743":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-25 12:57:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/69/754169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/69/754169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/69/754169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/69/754169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/69/754169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/69/754169/2"}}},"commit":{"parents":[{"commit":"4c2cc73d571dd86303bc16063951c13bec8af8c5","subject":"Add Centos-8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/4c2cc73d571dd86303bc16063951c13bec8af8c5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-09-24 16:56:11.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-25 12:57:18.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nDepends-On: https://review.opendev.org/754382\nChange-Id: I2c60a5ed861b3d8f364668ef9efb858a7cc895f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/be35f492ff030b6f03425639891a09bcf2619743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/be35f492ff030b6f03425639891a09bcf2619743"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
