)]}'
{"id":"openstack%2Fopenstack-ansible-os_horizon~824331","triplet_id":"openstack%2Fopenstack-ansible-os_horizon~master~I6b68d4d15ae516d23c88b3c4c21a076e8d54604e","project":"openstack/openstack-ansible-os_horizon","branch":"master","topic":"osa/include_vars","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-01-25 17:30:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b68d4d15ae516d23c88b3c4c21a076e8d54604e","subject":"Refactor use of include_vars","status":"MERGED","created":"2022-01-12 08:32:34.000000000","updated":"2022-01-25 17:34:12.000000000","submitted":"2022-01-25 17:30:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824331-osa/include_vars","meta_rev_id":"6a3ce38fcaa50f4473ba37612bd3aa3f989ce262","_number":824331,"virtual_id_number":824331,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-25 17:30:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2022-01-14 08:55:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2022-01-25 15:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2022-01-25 15:23:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 11:05:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-14 08:55:36.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2022-01-25 15:23:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"e3bca43a5cdb67c5e5c9aec08fc21c0fc4eb8b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-12 08:32:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7658915d1090b5a10ec14b3eb40c97a040bf8020","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 11:05:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-8 https://zuul.opendev.org/t/openstack/build/42f2b558569b4a37a85796f681b8e4ec : FAILURE in 1h 40m 39s\n- openstack-ansible-deploy-aio_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d35213699fd441c8c8aed1750a7f52d : SUCCESS in 2h 06m 29s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/deb025159f8f4490a61f703a7c1340c7 : SUCCESS in 1h 51m 18s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66d44a5ef518436a8cb908abb164fc67 : SUCCESS in 2h 09m 52s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/25d6f663bce544d6a617f622a30c83db : SUCCESS in 1h 18m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8a7d9fab314c49c58fc7950a5eb67853 : SUCCESS in 1h 28m 44s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7b5398ed9c44c188afd3408e53b2c04 : SUCCESS in 1h 29m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb8a87f0c02d48909a6dba289110f431 : SUCCESS in 1h 34m 38s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/83a60855bc244ccda38ff79dffeb659a : SUCCESS in 2h 20m 22s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f092a46941804a51adf7e1628c9a18b8 : SUCCESS in 2h 21m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/a3da4c5a1dfd423982ae5fc15705ddc8 : FAILURE in 35m 07s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b99ce7ac0a104fe399a5343449b8df73 : SUCCESS in 1h 03m 29s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/470b96b8f6f747f1aaeca8c7be774cde : SUCCESS in 22m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c31d5c933144f381e1805fdb896a69 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"59d0276774e4e643b6aaa628c737045d53a7d2c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 11:31:05.000000000","message":"Topic set to sa/include_vars","accounts_in_message":[],"_revision_number":1},{"id":"5c302142ed6c9f4293849baf752b1f3c73c70a6b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 11:31:10.000000000","message":"Topic sa/include_vars removed","accounts_in_message":[],"_revision_number":1},{"id":"3f464f08f6bc37c767fc6778131243dab379e5ff","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-12 11:31:15.000000000","message":"Topic set to osa/include_vars","accounts_in_message":[],"_revision_number":1},{"id":"0c984574c06e451ebabd1931aeba4904ca154022","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-01-13 13:13:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ce166d18358ac66d861d447652631ada301f818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 16:03:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/5ff78634b8e84d65a6d49f05cfb74e64 : SUCCESS in 1h 59m 13s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/dafc7dceb9f749b081e961393ed05ab3 : SUCCESS in 1h 59m 36s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/028cacb0ab894eca9c22c95adbc79c5b : SUCCESS in 2h 01m 21s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/c1513c960eaf4c62bd4d2e44409880c4 : SUCCESS in 1h 13m 34s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b47524dc2a443ed9a255227c11156d3 : SUCCESS in 1h 23m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56ceb2d22a8f4fd48f444d13f2bf87d2 : SUCCESS in 1h 38m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/acf0f08188ea415ca744c8630e4a20cd : SUCCESS in 2h 15m 56s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b4764b1945444a9b2a5f8df82245ac0 : SUCCESS in 2h 19m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/68133e7690a54ad5870ec1aefbf67c34 : FAILURE in 31m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cd1679cbb1b400aa179f65da7234b66 : SUCCESS in 58m 27s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/245f1e67836b408a8edd6ebeef52a785 : SUCCESS in 25m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5c0d324b88487cbab25c147fee5bfe : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"706d7345e2855f045212d3299340ca08ee00ac92","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2022-01-14 08:55:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eeb7b7e0c281d88ad0930114acd0963ba15d2167","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-01-25 15:23:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b75cd435df168460a20864478b67e274c89c51db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 15:26:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29a4433b95d9dc161e46d4a3be68d06e4665d5f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 17:30:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed6776c215f948819d3b802b7ead9174 : SUCCESS in 1h 57m 33s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/fab52121b6da4775a6c0b01b1ec80912 : SUCCESS in 1h 51m 48s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eac09789e73a4765b90d7c5d45383ca7 : SUCCESS in 1h 56m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/0313c20fb8784a52af7f1d36a853cf3c : SUCCESS in 1h 08m 26s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/c45c18c223ac40bcad98ed9c456d6a8f : SUCCESS in 1h 25m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8742bc99705e48dcaf1e917f54454632 : SUCCESS in 1h 27m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a1eb52333f14afba70c90aca37eef63 : SUCCESS in 49m 26s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d216ebe22ad45479f4e1e376304eaf9 : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd8d7013ad446539afa5ce6d27e5cbb : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4db4f3b93292a89ef617b7339d21e773480c8727","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 17:30:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a3ce38fcaa50f4473ba37612bd3aa3f989ce262","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-25 17:34:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3994d1110c86444394f28343a44b41ed : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c22a552c6f27e7e4bb06a4b2c9d55fbe35edba19","revisions":{"c22a552c6f27e7e4bb06a4b2c9d55fbe35edba19":{"kind":"REWORK","_number":1,"created":"2022-01-12 08:32:34.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/31/824331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/31/824331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/31/824331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/31/824331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/31/824331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/31/824331/1"}}},"commit":{"parents":[{"commit":"9b8cdbc525df71ea12af727c43e3dfb4b897d745","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/9b8cdbc525df71ea12af727c43e3dfb4b897d745"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 08:32:22.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-01-12 08:32:22.000000000","tz":0},"subject":"Refactor use of include_vars","message":"Refactor use of include_vars\n\nUse a first_found lookup instead of a with_first_found loop so that\nthe \u0027paths\u0027 parameter can be used.\n\nThis ensures that only vars from the role are included, and not vars\nfrom a parent calling role. This can happen when a parent role has\na higher priority vars file available for inclusion than the role\nit calls.\n\nChange-Id: I6b68d4d15ae516d23c88b3c4c21a076e8d54604e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/c22a552c6f27e7e4bb06a4b2c9d55fbe35edba19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/c22a552c6f27e7e4bb06a4b2c9d55fbe35edba19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
