)]}'
{"id":"openstack%2Fopenstack-ansible-os_horizon~847653","triplet_id":"openstack%2Fopenstack-ansible-os_horizon~stable%2Fyoga~Ia9fa3341dfa1386501bf74a558e418414a58fd1f","project":"openstack/openstack-ansible-os_horizon","branch":"stable/yoga","topic":"fix_allowed_hosts-stable/yoga","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-06-28 18:03:05.000000000","reason":"Change was submitted"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2022-06-28 18:03:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9fa3341dfa1386501bf74a558e418414a58fd1f","subject":"Fix ALLOWED_HOSTS","status":"MERGED","created":"2022-06-26 06:55:24.000000000","updated":"2022-06-28 18:04:22.000000000","submitted":"2022-06-28 18:03:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847653-fix_allowed_hosts-stable/yoga","cherry_pick_of_change":844815,"cherry_pick_of_patch_set":3,"meta_rev_id":"121102791c5e0a77a6ffb22431f70ce531d4b978","_number":847653,"virtual_id_number":847653,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-28 18:03:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2022-06-28 15:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-28 15:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2022-06-28 15:44:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"CC":[{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-26 06:55:24.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":34653,"name":"Danila Balagansky","email":"dbalagansky@me.com","username":"dbalagansky"},"state":"CC"},{"updated":"2022-06-26 09:28:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-28 15:39:00.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2022-06-28 15:44:45.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"}],"messages":[{"id":"2e7951fea6b7b8c008c4a23bdb30c3063c8992bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-26 06:55:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"387028ea22e9418bec520add3ec8f212b0b7ff74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 09:28:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/78abf888c1204874a5b4db4b7dcc97d9 : SUCCESS in 1h 57m 08s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/54ff3e92433743d2ac0800140b8aa29c : SUCCESS in 1h 55m 09s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/abfbb707c7224b528945cc6764c2559a : SUCCESS in 1h 44m 57s\n- openstack-ansible-deploy-aio_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/e6b16e24c40a48d4bc78e83045b82312 : POST_FAILURE in 2h 21m 09s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54a64e1c40dd425791a686d9b7271067 : POST_FAILURE in 2h 04m 50s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a83729bcee04fa4ad61cb50521f02f7 : SUCCESS in 1h 57m 09s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/6b9f2f6dde2141daa532ab9a5d46bb4c : POST_FAILURE in 1h 45m 33s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9e09fd6f46048d6a75476418db090c6 : SUCCESS in 1h 33m 00s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/b357fa0f0e764293a0edc6a94aaca4af : SUCCESS in 1h 21m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/693222064a62497b9e9a701937424e05 : SUCCESS in 1h 36m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65e9c5ac5fdf4d7ea5f34a52020cd7b6 : POST_FAILURE in 1h 46m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/b749e6bb0bc64cc993a3f8cd2b7cc3cb : SUCCESS in 1h 26m 43s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5791647c1ea4e1dae73ae6bdf268831 : SUCCESS in 2h 24m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de6c935269c14ea7898148f6d8f3ca5b : SUCCESS in 2h 13m 52s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/14a04de7311a46dbbda026574dc93973 : FAILURE in 56m 41s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5788e6bd9034087b27a68c6e4006e4d : SUCCESS in 51m 14s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4b46d8767d945ab903f49c4956553b7 : SUCCESS in 14m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f083d40b57e345b6a42a6ceeed865bea : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1cf9672855368358e533a0fe15e0bbb94e88d943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-27 17:50:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e95e08758f2ff66f766c4f69c8a601a8541cea2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 20:19:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/4642b144b6ae40dea50bb27ad3805631 : SUCCESS in 2h 12m 37s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec9880d475f54be698dfffb981362cf1 : SUCCESS in 2h 09m 43s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/633170e3ff3a4f47a1b57d85eb1847d2 : SUCCESS in 1h 48m 08s\n- openstack-ansible-deploy-aio_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ec2e631a356941c887d668aeddadc3d8 : SUCCESS in 2h 19m 58s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52690319256e4de5b7b6fa85b2da5bda : SUCCESS in 2h 00m 36s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/758ae61bf1374d238b8a825894242719 : SUCCESS in 2h 04m 23s (non-voting)\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/d90a46d62a0342ebb3c6a20c866e5e31 : SUCCESS in 1h 06m 18s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a76a2b3bf4f47468d7e6dedf1816767 : SUCCESS in 1h 33m 35s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/181f8086be004d81afe7f902c2a9132f : SUCCESS in 1h 25m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4b235f4d474489bab9978c006450a4a : SUCCESS in 1h 26m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8257f82918334a6abf26a8262efbbec6 : SUCCESS in 1h 22m 18s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/6b48e4c6bc37497c8a6a8a469e2f676f : SUCCESS in 1h 34m 07s\n- openstack-ansible-upgrade-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/73ab55e0c6c9419084fb957c5283be82 : SUCCESS in 1h 50m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/969499bd20f7482c81e561ff3a7cc8a7 : SUCCESS in 1h 47m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/38edd0834d854d36aa854888ac1ccb3a : FAILURE in 59m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09b463c398114882b4b926fa4b4125e6 : SUCCESS in 1h 02m 18s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa7ec4f9f0c5412f9b8e7c48bf2a4e83 : SUCCESS in 21m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d8cda717534c71b3ecebf328a85845 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"925c79520720dcf5f1c5aa7f4954f548c56b7ca2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-28 15:39:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3cacad55a68a2b43ca415c76032fba8ac697a85b","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2022-06-28 15:44:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"333ba1e7f96f4fb6d2fae7858cc4d3e52d678c9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 15:47:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"da3698dcee535e750f979821d224b383a1ca9b7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 18:03:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-deploy-aio_lxc-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad07d58149a84797a7fb377ca8b6fa1e : SUCCESS in 2h 05m 11s\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/b2aaa2c188cf4aee9134a3cd578e8494 : SUCCESS in 1h 51m 55s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/7c103f2e6f904ef7bf6783f15d0dcbb4 : SUCCESS in 1h 37m 58s\n- openstack-ansible-deploy-aio_lxc-rockylinux-8 https://zuul.opendev.org/t/openstack/build/a92c67b4ceac46f2b37271ad45199cc7 : SUCCESS in 2h 15m 41s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f221d2f56bd4210b96cf9a074b3eea2 : SUCCESS in 1h 52m 27s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/05135bb3f4b149d5a2faa16d77b06c90 : SUCCESS in 1h 05m 31s\n- openstack-ansible-deploy-aio_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/07af8b45987246c5a28683ae4ac06004 : SUCCESS in 1h 15m 44s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/50257c9c0d3d47e691e64be4da6cb964 : SUCCESS in 1h 09m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab15acc4179e4ea0a90decd0750efa4e : SUCCESS in 1h 26m 57s\n- openstack-ansible-deploy-aio_metal-rockylinux-8 https://zuul.opendev.org/t/openstack/build/ced6e3d70dea49729929c4cd33348f2a : SUCCESS in 1h 16m 53s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41b46ac8cb5b49a49db7b01c2ca02f2c : SUCCESS in 52m 07s\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a52aac9bec04e4b9985aea184a6726b : SUCCESS in 12m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ca1c6d727c4440288c9143f03a33308 : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"317e5f696051f79d06500a4cbea8425900a9debe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 18:03:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"121102791c5e0a77a6ffb22431f70ce531d4b978","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 18:04:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acdfca970a0d4326bf06bb1fd78b0b56 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a4fdc4b7071acab41c348d04ed1bc0f2f41492c3","revisions":{"bde526ca48ab5b2d4714e705a15a700114b1b2e9":{"kind":"REWORK","_number":1,"created":"2022-06-26 06:55:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/53/847653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/53/847653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/53/847653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/53/847653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/53/847653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/53/847653/1"}}},"commit":{"parents":[{"commit":"8ab99ba03cbc57aec099e420dce679ddfaa89c42","subject":"Update .gitreview for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/8ab99ba03cbc57aec099e420dce679ddfaa89c42"}]}],"author":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-06-06 09:26:10.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-06-26 06:55:24.000000000","tz":0},"subject":"Fix ALLOWED_HOSTS","message":"Fix ALLOWED_HOSTS\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-haproxy_server/+/846473\nChange-Id: Ia9fa3341dfa1386501bf74a558e418414a58fd1f\n(cherry picked from commit 7db1f112ef734999e9d884df641b7cab0c953e1c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/bde526ca48ab5b2d4714e705a15a700114b1b2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/bde526ca48ab5b2d4714e705a15a700114b1b2e9"}]},"branch":"refs/heads/stable/yoga"},"a4fdc4b7071acab41c348d04ed1bc0f2f41492c3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-27 17:50:11.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/53/847653/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/53/847653/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/53/847653/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/53/847653/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/53/847653/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/53/847653/2"}}},"commit":{"parents":[{"commit":"8ab99ba03cbc57aec099e420dce679ddfaa89c42","subject":"Update .gitreview for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/8ab99ba03cbc57aec099e420dce679ddfaa89c42"}]}],"author":{"name":"Danila Balagansky","email":"dbalagansky@me.com","date":"2022-06-06 09:26:10.000000000","tz":180},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-06-27 17:50:11.000000000","tz":0},"subject":"Fix ALLOWED_HOSTS","message":"Fix ALLOWED_HOSTS\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-haproxy_server/+/846473\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/847668\nChange-Id: Ia9fa3341dfa1386501bf74a558e418414a58fd1f\n(cherry picked from commit 7db1f112ef734999e9d884df641b7cab0c953e1c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/a4fdc4b7071acab41c348d04ed1bc0f2f41492c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/a4fdc4b7071acab41c348d04ed1bc0f2f41492c3"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
