)]}'
{"id":"openstack%2Fopenstack-ansible-os_horizon~888917","triplet_id":"openstack%2Fopenstack-ansible-os_horizon~master~Ida72c712153dcda4cd06e0959f98ade4fee8dfbd","project":"openstack/openstack-ansible-os_horizon","branch":"master","attention_set":{},"removed_from_attention_set":{"32666":{"account":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"last_update":"2023-08-22 22:25:23.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2023-08-22 17:15:21.000000000","reason":"\u003cGERRIT_ACCOUNT_28619\u003e replied on the change","reason_account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}},"hashtags":[],"change_id":"Ida72c712153dcda4cd06e0959f98ade4fee8dfbd","subject":"Fix wrong neutron_ml2_drivers_type","status":"MERGED","created":"2023-07-19 12:19:00.000000000","updated":"2023-08-22 22:26:17.000000000","submitted":"2023-08-22 22:25:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888917","meta_rev_id":"a238839ef17807c88f543092842ea0cc6614399c","_number":888917,"virtual_id_number":888917,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 22:25:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"all":[{"value":2,"date":"2023-08-21 19:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-21 19:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2023-08-21 19:55:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":1,"date":"2023-08-21 19:55:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-19 15:13:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-19 15:52:17.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2023-07-19 15:52:24.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"52df910cabae6446ab6e496c2eb1481b702b479e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-19 12:19:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"632bf901d7b53a3cd4dcbced372e3cd21c55d2f4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-07-19 12:19:30.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"45664c4d810cd3a28a84f52347c52e5d6637fdbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 15:13:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6ceb86884ff44c19daa94a3142a2ce6\n\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ad7bb57fd3340a98b781a197a289987 : FAILURE in 44m 34s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/55f8572a929e43179d82f6347237d58f : SUCCESS in 1h 57m 04s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/d47d29ac58f1438eb758196ba6e55b35 : SUCCESS in 2h 05m 55s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcc47c370f304b61b961eae724365bf8 : SUCCESS in 1h 57m 54s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ee7dc63968af4a05a4ead8d44312fff4 : SUCCESS in 1h 31m 39s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/4892eed10c284f65885f9e970f1518b5 : SUCCESS in 1h 34m 40s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/a74d7f76c8eb40ecb51a3553cf781925 : SUCCESS in 1h 42m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eaa69f90ea304106b21683652976b563 : SUCCESS in 1h 36m 58s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ac5b6e47e91e4896a5a34a4940b1ee2e : SUCCESS in 1h 49m 34s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fbd5aaa2f5b4d0fbe14d394d11fc8d3 : SUCCESS in 1h 43m 16s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/610696f77f2a433da5e17b40367ad112 : SUCCESS in 2h 47m 23s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad32a1f0c89040b59f748ef11a0a6177 : SUCCESS in 2h 35m 51s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7ed972ba8d644cf7ae47f48e2203d1c7 : FAILURE in 55m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29b982abee374e64ab4fd95743727e60 : SUCCESS in 1h 02m 07s (non-voting)\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b7d8ab17bdf4a8fbd30a6e3316bc888 : SUCCESS in 11m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3fbb78e6f1b4a4fab9f4cec497b9c6a : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"aa507b7ea093f7b92abc5864a59bd2523dede769","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-07-19 15:52:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dd3b6f108b669ef70fcfcf3b7280ba814aafff29","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2023-07-24 16:30:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4806ba14b4a4966c9236b9a0a83f27896b61668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 16:33:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40dd1053ffe89dc71d6bdfe4f67ddd97ccae2931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 18:54:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c828291feb3426ebc31df977482701d\n\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/67ee25d8edfa497fb04b61a72438f112 : SUCCESS in 2h 15m 46s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/697bfb5307a24dd5b034d81df500a219 : SUCCESS in 1h 49m 05s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/192e08098899462ba69f264f63744aad : FAILURE in 20m 50s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/5a8880bab7a84e6f8e92ee64f948aacb : SUCCESS in 1h 30m 54s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7449c92a6b55499ca86087b22bd6f41e : SUCCESS in 1h 50m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/183cb2d2368f4dd582bd4ba0e948cf88 : SUCCESS in 1h 27m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/77444bb11b4945018e9af90c34a5bde3 : FAILURE in 1h 30m 18s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23d2e2b81c4e46578c793066763993dd : SUCCESS in 2h 00m 54s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7770ae424b86448f82b10d74bda9239a : SUCCESS in 14m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52eb39e18654708991dfb9b2f867beb : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"64a3c7fa5dec661d1d5d41c177556a735236845e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-15 14:30:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1922157739ab2589a14d29ba5e2ca6be7889788b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 20:43:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b16937455af49249cfb30170b5de635\n\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/35c10dae357147bd92843bea97c14ad9 : SUCCESS in 2h 11m 00s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/d46bdc0902fc482e8af576e6c9b03133 : SUCCESS in 1h 57m 28s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/10012920331d4230b18db8adbe48f53f : SUCCESS in 2h 00m 00s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7d6eb8b2a3447f99b29ee871409641f : SUCCESS in 1h 48m 44s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/bef8f70d88d34ca1983f5cfe342343e3 : SUCCESS in 1h 41m 30s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a9b45ea273749d896d0d74d206fe0b3 : SUCCESS in 1h 40m 07s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/07f73baf83a1463aafcc67dcfe14581d : SUCCESS in 1h 45m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef471bdcdb174bc88018c42b6e5bf92c : SUCCESS in 1h 35m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/fa2c8451a8b343d289e5dfd874361805 : POST_FAILURE in 1h 54m 24s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25145f9083b14c71862daa7d21ddae28 : SUCCESS in 1h 44m 08s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/45a06a6b98dc4e9297032dd308d114d7 : SUCCESS in 3h 00m 27s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c1c8e9c149e48e495defe918c0dc82a : SUCCESS in 2h 12m 24s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c7e3a6e2a6f24898aadbbd9264257f89 : FAILURE in 54m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/292a8ab9073d44a9bc69e1d6fc1873c0 : SUCCESS in 54m 43s (non-voting)\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d314cc4c43da4ee3b7895c8b54059be2 : SUCCESS in 17m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873372c71d7c41e995a79b864266bd9a : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"358aedf9e86f8aac6a941537c82782905e650abd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 19:55:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"426b974f3f75b73c8eb489b371124aa18560d641","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 19:55:31.000000000","message":"Patch Set 2: Backport-Candidate+1 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9454c0abb53033560c27d411c85114e835a5733c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 22:36:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98633c158fc8437fad10f6fd5723157b\n\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/a81af950c01f40c09f5e84b5883bdd7c : SUCCESS in 2h 03m 11s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/911ede598dc147f1a8b04e17fe84c2c2 : SUCCESS in 2h 17m 47s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/600a76d827784d41874fc6049d653165 : SUCCESS in 1h 48m 35s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/008511dc362049f2af9134262e2e3dd3 : SUCCESS in 1h 50m 53s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/57a4389c71c54d4a9de70fb6e85da625 : SUCCESS in 1h 16m 24s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/c123e432777145038ee4bab1b9e48b24 : SUCCESS in 1h 40m 58s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/cf74b8830e9b4383aec56c58990c4edf : SUCCESS in 1h 42m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/499010a34705469a9419c2d046c07490 : SUCCESS in 1h 27m 23s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/15d2cff954324370bcc3f7d45482f14d : SUCCESS in 1h 35m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f828bccd7a1346a780f9f66278187262 : SUCCESS in 1h 29m 31s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/3e43bcd1232d476b8979fee431e46363 : FAILURE in 2h 17m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fadb68c4335544f588247e3d13a663b3 : FAILURE in 2h 31m 46s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/2116298358e844838d459fc87a338964 : FAILURE in 49m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/189faf6d7dfd4c36b8132c9fa7b193a5 : SUCCESS in 54m 58s (non-voting)\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7e174a5dd724471b95b5ac038ecc583 : SUCCESS in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b15912c08b45258970235543d0cb17 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3c2249c0bf78427b9f93e2af9ecf195ba22372f4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-22 17:15:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdfc196ba2f9e86436cf49021c1c70485604ba2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 19:58:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89a5a6333c924f53b26b712b037e9323\n\n- openstack-ansible-deploy-aio_lxc-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d5775e7c712472ebda55ef52f9c2339 : SUCCESS in 2h 00m 04s (non-voting)\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/31b25a20c35f442d86d90272ef370827 : SUCCESS in 1h 57m 49s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/011570d319da4aeebd1279923e6e10f8 : SUCCESS in 1h 53m 44s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4638c1f114034058986c48bf3725716d : SUCCESS in 2h 00m 47s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/532a778181c141f893c7333642de1fb4 : SUCCESS in 1h 26m 16s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd6760ebcca54cefb96c679f85e0b839 : SUCCESS in 1h 33m 32s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/0d4227fc69c448e7be602dcabe7ed30b : SUCCESS in 1h 33m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d890edcfa5f4713ac65efd43dc83c7f : SUCCESS in 1h 31m 21s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/f9a6af2323fc44029655affaba0d750a : SUCCESS in 1h 26m 44s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ec0ec3c58d045f2bf2c242da54e986b : SUCCESS in 1h 33m 19s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/7e96337f5d0a4ea7b9e184a27c620b21 : SUCCESS in 2h 37m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a40cbbe62a94727befda638ec6cd00c : SUCCESS in 1h 52m 09s\n- openstack-ansible-deploy-aio_distro_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/6699325fd4a2457e99fb3e51c1be5805 : FAILURE in 57m 14s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80d94904d8be46afbe8df1ea5f088de7 : SUCCESS in 58m 53s (non-voting)\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a87b37c23f441c492a826b52fb57808 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9ef53e68a64778877d2446bbb20dd9 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"810d0f6ed8c7a5af5e5d416ea3365688ecb44941","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 20:00:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"83fe29549b83a3de2a5ff8fd2ec0427578f521cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 22:25:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b70979a8238e41079a93530589368c0e\n\n- openstack-ansible-deploy-aio_lxc-rockylinux-9 https://zuul.opendev.org/t/openstack/build/e4118a98488a49179429999c1250212b : SUCCESS in 2h 18m 09s\n- openstack-ansible-deploy-aio_lxc-debian-bullseye https://zuul.opendev.org/t/openstack/build/4c6807eb1d134c1a957e280dfec66c28 : SUCCESS in 1h 52m 27s\n- openstack-ansible-deploy-aio_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ccf75d35bd9492c90f78ad741f32107 : SUCCESS in 1h 51m 05s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/2b570d6d458244a69e63959263891931 : SUCCESS in 1h 21m 42s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/9be7daf7c4114e10b6416d69d6ca19fd : SUCCESS in 1h 45m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/261792147fd54893961432a77738db89 : SUCCESS in 1h 24m 30s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/c9460fc7ca534790b3f64d3da3bb8f05 : SUCCESS in 1h 40m 32s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/752323863faa4fedbd07ff833ff4a3bb : SUCCESS in 1h 25m 08s\n- openstack-ansible-linters-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1303d43bbc2b461a99743d6d265d893f : SUCCESS in 10m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50359fe9ef7b439a964a1274ae42ad16 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"41aa0e4d34145b0d0297c2b188f0784514035fc0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 22:25:23.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"a238839ef17807c88f543092842ea0cc6614399c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 22:26:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b856c961cd01430885bbe90942478038\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f2398e51f65458f8db4e5ca5c749857 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5d635c469af4e9c9bb85615257e3bb3738ce3a29","revisions":{"3c6e7f32c47b2731d4f2036db77a51cbf7742d8c":{"kind":"REWORK","_number":1,"created":"2023-07-19 12:19:00.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/888917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/17/888917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/17/888917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/17/888917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/17/888917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/17/888917/1"}}},"commit":{"parents":[{"commit":"db7110d29788701a23ee8713a58c70e4e8e01ad2","subject":"Add PKI support to horizon backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/db7110d29788701a23ee8713a58c70e4e8e01ad2"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-19 12:18:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-19 12:18:53.000000000","tz":120},"subject":"Fix wrong neutron_ml2_drivers_type","message":"Fix wrong neutron_ml2_drivers_type\n\nneutron_ml2_drivers_type is a string, not a list, while default\nvalue for it is a list. Local settings also assume that it is a list.\n\nSo we ensure that the falue is a string by default now and\ntreat it as string in the code.\n\nChange-Id: Ida72c712153dcda4cd06e0959f98ade4fee8dfbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/3c6e7f32c47b2731d4f2036db77a51cbf7742d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/3c6e7f32c47b2731d4f2036db77a51cbf7742d8c"}]},"branch":"refs/heads/master"},"5d635c469af4e9c9bb85615257e3bb3738ce3a29":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-21 19:55:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/17/888917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_horizon","ref":"refs/changes/17/888917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/17/888917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/17/888917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/17/888917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_horizon refs/changes/17/888917/2"}}},"commit":{"parents":[{"commit":"1ebeab7d26a32a9f29b680b59e526d57d8371ea4","subject":"Merge \"Stop reffering _member_ role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/1ebeab7d26a32a9f29b680b59e526d57d8371ea4"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2023-07-19 12:18:53.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2023-08-21 19:55:26.000000000","tz":0},"subject":"Fix wrong neutron_ml2_drivers_type","message":"Fix wrong neutron_ml2_drivers_type\n\nneutron_ml2_drivers_type is a string, not a list, while default\nvalue for it is a list. Local settings also assume that it is a list.\n\nSo we ensure that the falue is a string by default now and\ntreat it as string in the code.\n\nChange-Id: Ida72c712153dcda4cd06e0959f98ade4fee8dfbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/5d635c469af4e9c9bb85615257e3bb3738ce3a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_horizon/commit/5d635c469af4e9c9bb85615257e3bb3738ce3a29"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
