)]}'
{"id":"openstack%2Fopenstack-ansible-os_ironic~654516","triplet_id":"openstack%2Fopenstack-ansible-os_ironic~stable%2Fstein~Icbb67998ef407d35d4b08f265082fc300d568e62","project":"openstack/openstack-ansible-os_ironic","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Icbb67998ef407d35d4b08f265082fc300d568e62","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-22 15:38:31.000000000","updated":"2019-04-22 20:16:45.000000000","submitted":"2019-04-22 20:16:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654516-1555964205229-354c625b","meta_rev_id":"0b42f8a82e411c114ac5acd1fdf4bff5b07f5613","_number":654516,"virtual_id_number":654516,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-22 20:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-22 16:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-22 16:11:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 16:11:16.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-22 20:16:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee8e09139cd0bfd3fd74f160eb5774aeb4053341","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-22 15:38:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88bd4b4563512bf5b154fe62facf7dba11459b16","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:11:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"446b7a57d3820154c1cc5e2b7654c057871f6dd0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 16:22:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/16/654516/1/check/openstack-ansible-linters/6883216/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/16/654516/1/check/openstack-tox-docs/a71674b/html/ : SUCCESS in 2m 53s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/16/654516/1/check/openstack-ansible-deploy-aio_metal_ironic-debian-stable/ab709a6/ : FAILURE in 21m 08s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/16/654516/1/check/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/acb88cc/ : FAILURE in 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2d64824826c1be8f3ee3d66885be43f24b25f194","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:45:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea320c276bfb44d01fed731e982fe8ed6a8418e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:03:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/16/654516/1/check/openstack-ansible-linters/68980d8/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/16/654516/1/check/openstack-tox-docs/1951198/html/ : SUCCESS in 2m 44s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/16/654516/1/check/openstack-ansible-deploy-aio_metal_ironic-debian-stable/99e8de8/ : SUCCESS in 2h 06m 16s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/16/654516/1/check/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/8188a97/ : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2677ee3499548c5c898fc5cd9e6244bacfb2d6a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 19:03:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b772488e01134815ccdc25cefaa3888c62014dcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:16:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/16/654516/1/gate/openstack-ansible-linters/4839f38/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/16/654516/1/gate/openstack-tox-docs/2e2ea3d/html/ : SUCCESS in 2m 46s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/16/654516/1/gate/openstack-ansible-deploy-aio_metal_ironic-debian-stable/ebbc2b0/ : SUCCESS in 1h 03m 57s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/16/654516/1/gate/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/f3e154f/ : SUCCESS in 1h 06m 56s","accounts_in_message":[],"_revision_number":1},{"id":"61f0e4a3de301e760138d6878a098d5d5cba08d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 20:16:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e12e8595e6e5858e5a492b7604472d52505374cf","revisions":{"e12e8595e6e5858e5a492b7604472d52505374cf":{"kind":"REWORK","_number":1,"created":"2019-04-22 15:38:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/654516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/16/654516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/16/654516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/16/654516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/16/654516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/16/654516/1"}}},"commit":{"parents":[{"commit":"1e68fb2ed73cfbe275ae15b6a5bffb798ac6a7a6","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/1e68fb2ed73cfbe275ae15b6a5bffb798ac6a7a6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:38:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-22 15:38:29.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Icbb67998ef407d35d4b08f265082fc300d568e62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/e12e8595e6e5858e5a492b7604472d52505374cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/e12e8595e6e5858e5a492b7604472d52505374cf"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
