)]}'
{"id":"openstack%2Fopenstack-ansible-os_ironic~654908","triplet_id":"openstack%2Fopenstack-ansible-os_ironic~stable%2Fstein~I5e89f0abec2f3ab1d04fe4d6ea83e6d3982cc0bb","project":"openstack/openstack-ansible-os_ironic","branch":"stable/stein","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I5e89f0abec2f3ab1d04fe4d6ea83e6d3982cc0bb","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-04-23 09:24:07.000000000","updated":"2019-04-24 08:46:28.000000000","submitted":"2019-04-24 08:46:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654908-1556095588473-332a2ac9","meta_rev_id":"944f84654cdc60f5075124e9deb161333fcff90f","_number":654908,"virtual_id_number":654908,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-04-24 08:46:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-23 20:23:33.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2019-04-23 10:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2019-04-23 10:49:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-23 10:49:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-04-23 20:23:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-04-24 08:46:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91a2b9f209fbef291031e69390dd06d148cb609a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-04-23 09:24:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b2a5b707cd24bb3704667218914d8cc151cb336","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:40:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/08/654908/1/check/openstack-ansible-linters/4ddd5c5/ : SUCCESS in 4m 15s\n- openstack-tox-docs http://logs.openstack.org/08/654908/1/check/openstack-tox-docs/95aa5ba/html/ : SUCCESS in 2m 35s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/08/654908/1/check/openstack-ansible-deploy-aio_metal_ironic-debian-stable/c4ded3f/ : SUCCESS in 1h 03m 23s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/08/654908/1/check/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/81c05d6/ : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f56328119af971d3f0f1b0f592f312354b96e99c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-23 10:49:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2e636ac1a34e47826e1c2839b7cfbeeff4ee365","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 10:50:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2bb3e9bd942ca9c3f8886cf6c1d0cf27c00d7c27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:10:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/08/654908/1/gate/openstack-ansible-linters/d881789/ : SUCCESS in 5m 09s\n- openstack-tox-docs http://logs.openstack.org/08/654908/1/gate/openstack-tox-docs/ad7570b/html/ : SUCCESS in 3m 00s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable finger://ze02.openstack.org/236462a8ca7c4015857617c33da5e816 : RETRY_LIMIT in 37m 34s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic finger://ze11.openstack.org/c7d919bbb6f5483a83e17aa34beb71fc : RETRY_LIMIT in 42m 08s","accounts_in_message":[],"_revision_number":1},{"id":"934cdb5f42feff67c16ed6f67e7ab7b10a2a2582","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-04-23 20:23:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bbfa30693dc22eab88be1ec990a8297a54c9351c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 23:07:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/08/654908/1/check/openstack-ansible-linters/579ebea/ : SUCCESS in 8m 41s\n- openstack-tox-docs http://logs.openstack.org/08/654908/1/check/openstack-tox-docs/06a5b01/html/ : SUCCESS in 2m 31s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/08/654908/1/check/openstack-ansible-deploy-aio_metal_ironic-debian-stable/e4bc6fe/ : SUCCESS in 1h 09m 27s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/08/654908/1/check/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/8aee644/ : SUCCESS in 1h 13m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7c0345a236670fec3114c50e5fce5aaddb16d044","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 23:08:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adb4eacbf90189fe9d2fda9777a34f8437bedafa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 00:33:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/08/654908/1/gate/openstack-ansible-linters/79b3f23/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/08/654908/1/gate/openstack-tox-docs/f216542/html/ : SUCCESS in 2m 28s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/08/654908/1/gate/openstack-ansible-deploy-aio_metal_ironic-debian-stable/11c395f/ : POST_FAILURE in 1h 21m 28s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/08/654908/1/gate/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/aad4529/ : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"02fff80777f1a9b7d17f28bf647dd1c755513575","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-24 05:52:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"17c13989d453bee25141ac0d2724bfcfce0fe9da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 07:35:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/08/654908/1/check/openstack-ansible-linters/d5d1725/ : SUCCESS in 6m 04s\n- openstack-tox-docs http://logs.openstack.org/08/654908/1/check/openstack-tox-docs/d712a49/html/ : SUCCESS in 2m 47s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/08/654908/1/check/openstack-ansible-deploy-aio_metal_ironic-debian-stable/aaf638b/ : SUCCESS in 1h 40m 27s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/08/654908/1/check/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/b01779f/ : SUCCESS in 1h 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"bdaae59f10999087b1c9f25205d9a4bf5ad13ffa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 07:35:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9f2d34cb929a933d174bb1b77c53ac40dc85a2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 08:46:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b52f3ae682e33dad9a93b65026c6abf26830df42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 08:46:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/08/654908/1/gate/openstack-ansible-linters/9d1cf76/ : SUCCESS in 6m 59s\n- openstack-tox-docs http://logs.openstack.org/08/654908/1/gate/openstack-tox-docs/76893b0/html/ : SUCCESS in 2m 30s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/08/654908/1/gate/openstack-ansible-deploy-aio_metal_ironic-debian-stable/a86f521/ : SUCCESS in 1h 00m 24s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/08/654908/1/gate/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/3f19ff5/ : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1b309288d0788ae0fd2034305b4c5ffb3212b57","revisions":{"b1b309288d0788ae0fd2034305b4c5ffb3212b57":{"kind":"REWORK","_number":1,"created":"2019-04-23 09:24:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/654908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/08/654908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/08/654908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/08/654908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/08/654908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/08/654908/1"}}},"commit":{"parents":[{"commit":"e12e8595e6e5858e5a492b7604472d52505374cf","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/e12e8595e6e5858e5a492b7604472d52505374cf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:24:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-04-23 09:24:07.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I5e89f0abec2f3ab1d04fe4d6ea83e6d3982cc0bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/b1b309288d0788ae0fd2034305b4c5ffb3212b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/b1b309288d0788ae0fd2034305b4c5ffb3212b57"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
