)]}'
{"id":"openstack%2Fopenstack-ansible-os_ironic~658024","triplet_id":"openstack%2Fopenstack-ansible-os_ironic~master~I12c47b03e466cad00569d3f741e3731ba9cec481","project":"openstack/openstack-ansible-os_ironic","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I12c47b03e466cad00569d3f741e3731ba9cec481","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-05-09 11:34:44.000000000","updated":"2019-05-09 14:16:32.000000000","submitted":"2019-05-09 14:16:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658024-1557411392390-a4b81bb3","meta_rev_id":"ba0089b59f3398d19bfa22278d1153a17f473187","_number":658024,"virtual_id_number":658024,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-09 14:16:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-05-09 12:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":1,"date":"2019-05-09 12:59:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 12:59:14.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-09 14:16:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6a17baaeb35b4923033fefc7c7619c66bcb1de2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-05-09 11:34:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f22c3988aa04896ac7e9d3345c40e6ff70b7e48","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-09 12:59:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37cebbdc8e1bf0c04e0938034346079d10c4e293","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:03:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/658024/1/check/openstack-ansible-linters/26f8c92/ : SUCCESS in 5m 41s\n- openstack-tox-docs http://logs.openstack.org/24/658024/1/check/openstack-tox-docs/894e8cb/html/ : SUCCESS in 2m 40s\n- build-openstack-releasenotes http://logs.openstack.org/24/658024/1/check/build-openstack-releasenotes/54305af/html/ : SUCCESS in 2m 29s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/24/658024/1/check/openstack-ansible-deploy-aio_metal_ironic-debian-stable/d32fc14/ : SUCCESS in 1h 12m 10s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/24/658024/1/check/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/c000050/ : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5bb70d92816e83cd27b070d778796e3fae9e4deb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 13:03:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f814ef494364d85f3518863101dc6416e97fb264","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:16:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"791e7810e610627be05af143b467dc2d05d96199","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 14:16:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/658024/1/gate/openstack-ansible-linters/283e666/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/24/658024/1/gate/openstack-tox-docs/7f32716/html/ : SUCCESS in 2m 29s\n- build-openstack-releasenotes http://logs.openstack.org/24/658024/1/gate/build-openstack-releasenotes/cf78642/html/ : SUCCESS in 2m 52s\n- openstack-ansible-deploy-aio_metal_ironic-debian-stable http://logs.openstack.org/24/658024/1/gate/openstack-ansible-deploy-aio_metal_ironic-debian-stable/3dabaca/ : SUCCESS in 1h 08m 17s\n- openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic http://logs.openstack.org/24/658024/1/gate/openstack-ansible-deploy-aio_metal_ironic-ubuntu-bionic/2c7d8a3/ : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64431f7483b5c2f503fc7a6f21bf3729b7d4885c","revisions":{"64431f7483b5c2f503fc7a6f21bf3729b7d4885c":{"kind":"REWORK","_number":1,"created":"2019-05-09 11:34:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/658024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/24/658024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/24/658024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/24/658024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/24/658024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/24/658024/1"}}},"commit":{"parents":[{"commit":"81d324836a2ce1e6cd7f9b30ac9fb456413be076","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/81d324836a2ce1e6cd7f9b30ac9fb456413be076"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:34:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-05-09 11:34:43.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I12c47b03e466cad00569d3f741e3731ba9cec481\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/64431f7483b5c2f503fc7a6f21bf3729b7d4885c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/64431f7483b5c2f503fc7a6f21bf3729b7d4885c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
