)]}'
{"id":"openstack%2Fopenstack-ansible-os_ironic~693281","triplet_id":"openstack%2Fopenstack-ansible-os_ironic~master~I9e0c48a6d6a6e6b128add994e4df075b334b8984","project":"openstack/openstack-ansible-os_ironic","branch":"master","topic":"fix","hashtags":[],"change_id":"I9e0c48a6d6a6e6b128add994e4df075b334b8984","subject":"Fix typos in yml files","status":"MERGED","created":"2019-11-07 03:13:41.000000000","updated":"2019-11-22 16:18:57.000000000","submitted":"2019-11-22 16:17:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693281-1574439441579-ca8dfbed","meta_rev_id":"67d858e14bbc0f989d1ff741a85edbccd9526d34","_number":693281,"virtual_id_number":693281,"owner":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-14 07:55:10.000000000","_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 16:17:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"value":2,"date":"2019-11-22 13:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-22 07:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"value":1,"date":"2019-11-22 13:22:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-22 07:56:48.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-22 13:22:38.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-11-22 16:17:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"421f2cfc617b9b1a3a14cbb9c2f43afa765b1119","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-11-07 03:13:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d14e765ea6d09587a8544f45dffcba3fa4e4950b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 06:09:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8ccef9e8ae6f4f47a6f132ff6b955276 : SUCCESS in 4m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8e8112b0eb7f4f88ae0b1dbe9f76d330 : SUCCESS in 1h 26m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a9ddd207e59d4dfbbcf5d6d972e3b7d2 : SUCCESS in 1h 47m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5e10a37fbeb7415ebeca6194d8d4b4d8 : SUCCESS in 1h 22m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a6642d53b8b4e2ab469f3ab49121b73 : FAILURE in 16m 18s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c9b39ec6adad429f8e729707a8102d65 : SUCCESS in 2h 52m 10s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1304b80155324971960e67c3f2f44bb9 : FAILURE in 1h 03m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cbccf80c8b849edbf88e18f96c6981e : SUCCESS in 4m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1422a052d901442e8e759810d70ac729 : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b6948454971827d008e95009e7e37163d18206bd","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-11-08 08:32:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4853c8f41b673190dc0233bec12cca3499257902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 11:41:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/567975b7ef874bf686eadf2f7a687fe2 : SUCCESS in 6m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/dcdca94d1efd47d7913c508071e9223d : SUCCESS in 1h 32m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7aaaa36c37d24ff39a5d74e24e00b3a4 : SUCCESS in 2h 01m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/929309e5ce33406486ab548bd46044c9 : SUCCESS in 1h 29m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab001c180d59496b950df26861e74296 : SUCCESS in 1h 19m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9c6faaacb3ba4130bd5e74074606c19b : TIMED_OUT in 3h 03m 21s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35b34358896d425d823c296e63d60284 : SUCCESS in 1h 42m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac7aaea95244f73ba2ab2db91ab59e0 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7885e6b041bf4ceaad66f23b3d304e33 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"a797323504d20a512d1ebbe52e352919ab1de9a6","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-11-11 00:34:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0fd09fbe034d1fd74b9bf69a17ffd56d88c2425b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 03:38:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f1d9e1de843c40868d965a7e582df718 : SUCCESS in 8m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/6d7738bff233426594b3fe46c111d9f1 : SUCCESS in 1h 29m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8d8c9943f5ae4663bcae3cf7defecdd6 : SUCCESS in 1h 46m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c0fc1a6a3bb6493d8c7ccfeb1372beea : SUCCESS in 1h 35m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ebe4ae7326b42849c6247cd863f92e1 : SUCCESS in 1h 16m 53s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0b36ddd2d2e64d9babc3b5c6720f622b : TIMED_OUT in 3h 03m 14s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0defcd677cb944649480fa6bb6f3ce49 : SUCCESS in 2h 06m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c172017fbac467ab4612e8c5bf9e155 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a3d76b37e0e4d10b921fbd008462ef6 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d9d3dab7a7f05cea5813cf7a67ac52183c3dbc60","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-11-12 03:11:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7e62c66831f95323a4a4cfe983291fa1fce57587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 06:16:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7f58b04e512c4befa12fd868c19597d7 : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3bec7235948a436bae47829a80c8388e : SUCCESS in 1h 35m 46s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3bcad4cc06a44512842d54f7bbd32d09 : SUCCESS in 1h 40m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2ad59981c4ec4686967809cf6be06dc9 : FAILURE in 23m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23b5fcaaa7da4369920544ec46e48a9d : SUCCESS in 1h 13m 43s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/842c64db7f3b46d0b320315d2f5fe725 : TIMED_OUT in 3h 03m 36s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c4fc1216d141412780b72ca9be7246ab : SUCCESS in 2h 05m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfdb3195a72041fa95c4407b3f6b11ba : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/146f835599584c4ba555fb15cc70c191 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4cfeaf9d220f174179755316dfb54f8764774450","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-11-14 07:55:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"725fa083381cee639a9e79bbfc9e652a10b1ca9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 11:00:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6c5e20b3042e4e68866c113d47bfae1e : SUCCESS in 5m 53s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9d7125192621406099d81b2ce723ac6d : SUCCESS in 1h 24m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5d3148344cf842fe94ea6729f424af2b : SUCCESS in 1h 59m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/711692f8f0ec40989852d87675300761 : FAILURE in 24m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/981b94628b604dccaaf61b0c711594c9 : SUCCESS in 1h 17m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c014382353dc4b23adb2c27d42e5dc26 : TIMED_OUT in 3h 03m 41s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a82a6fb6f2f4d09b549208d4534fab0 : SUCCESS in 2h 15m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b838748893ee471c97cecd14128dcf84 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ede2089f63ba478aab4dda847ad75e18 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a9a5415a83ed3b37517322243910cb970494591e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 10:21:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"012acfa83cd3e825c3773bea3f5fdac9c678bf41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 13:29:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dec09f9ab22b4c02afa2db8d3afdc19d : SUCCESS in 7m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c53b8d036fea4d4caa0fbf70814a70a6 : SUCCESS in 1h 25m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/db8d56843a89416b9189b6e6d40544e5 : SUCCESS in 1h 55m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9926f6a337fa4e63bc9806dac396f9c8 : SUCCESS in 1h 35m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4095f58c26e84b9085b2b4d6156743f6 : SUCCESS in 1h 27m 42s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ea81312444bf4a0886a354fd1d701397 : TIMED_OUT in 3h 02m 57s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/549c743d507140efba51408c1f36e855 : SUCCESS in 2h 13m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9efeb835c9b74708ac6b35336123142f : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6bbcdd936006490192c79020943df4fa : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0bede2796d17a8e06669b4fc1582f509637d24e6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 13:34:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"66efd09fc9062801075e6905d15e0ab8f559e627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 16:40:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/583b7d5a310644c7b37ec98931f9d87c : SUCCESS in 6m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/0f8eff429bd54de197a8543387265375 : FAILURE in 1h 38m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/835d8aca63af4e0ab0139787b3dfcef6 : FAILURE in 1h 59m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f0e3c7a853054f66925edcb4b29b1b4a : FAILURE in 1h 59m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/511c14762388497bb4e67e48b00f12d0 : FAILURE in 1h 31m 14s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/39eeac30870342b9af1a19e44e68f5d5 : TIMED_OUT in 3h 03m 59s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4e4699fd21548ad9af1a9cb589fd523 : FAILURE in 2h 32m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447bd7af14804e21945b1e3a06da7f99 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/388d2554f6804642a4e50f88ef87a942 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":1},{"id":"012e4c4bbb654ee262fe8989c80fbe37e02bdb49","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-20 09:28:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ddfdc0241c24edd82c34b8973ce8578dad6d84b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 13:04:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3900fc39c0a842ffa28989e9820bbe04 : SUCCESS in 5m 53s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/676d540cd4214bb1a3c9327dd8502605 : SUCCESS in 1h 33m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae86c08306b94972845e888b20356be8 : SUCCESS in 1h 42m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/31ad26966c474210a5c0eb6b8387079f : SUCCESS in 1h 54m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98a6934201334905aa39b7e2e5c36a22 : SUCCESS in 1h 19m 58s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/100aee3c68904b789cf2eace41ff661c : TIMED_OUT in 3h 02m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1039bc01250547a69f499c582fe43fff : SUCCESS in 2h 03m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/763ef51973714169848ee45945f369cd : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cce586325410407d885174c9eff4525e : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bad29624c4c9c3c78778706540c5c6f5ed96c40e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-20 13:11:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3ce22f1d93a1b5101fc49aaad70d1ae1fced6da6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 16:21:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/65d9edd8b20541cb9831abd034e45884 : SUCCESS in 4m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ff03c19226b648d4863b505b24e0bfb7 : SUCCESS in 1h 30m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fc0e5f425262419f9ee7564480380f65 : SUCCESS in 1h 45m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/08e89b7e32f444eaa305556e4eba5b8c : SUCCESS in 1h 32m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a473bfd9d9d14887938a3f05646b96e4 : SUCCESS in 1h 21m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b9b9c3eec8d8434eb1407ef17bc7aaa1 : TIMED_OUT in 3h 02m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d859c69a2b14750b7ed6ed0a4927779 : SUCCESS in 1h 59m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd25686e3974ac198c3953644715841 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06479321dd0d4705b279bb4c7858322f : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"d61d2b7610fe804c9ec05f6b524bb6892e75ccd5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-20 17:48:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cf7379c74854bb8ba9140135e55390007a4f16ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 20:56:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e5a10f5822b843b0b24b3d6b1b898613 : SUCCESS in 5m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/afb0304d356d400aa08b026fff9f6337 : FAILURE in 1h 21m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7e2801753fb54a59bed2d1baabbb6623 : SUCCESS in 1h 55m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2921ee866d1e4f3ea8b7945416e29abd : SUCCESS in 1h 33m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d155732d3e5f430dbbcc2d046a3c04de : SUCCESS in 1h 20m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d7aa87905a984ef98a559cb3f513fffc : TIMED_OUT in 3h 03m 17s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac2c06f13be344d185d5514246c48f36 : FAILURE in 1h 19m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43679959f69945a78a4cb23cf8f2fd57 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24b56ade721f4a8db00edeea3438219f : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9738fb35c9363cb6bb2495f8c8266903247d4650","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-20 22:17:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cbffc7f0600df2a96a1220a2e9dce609f7be41d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 01:25:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2d7207c92252444eb1840309ee95db7c : SUCCESS in 5m 58s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/eea871bb35f34a218290c71c507e0e7e : SUCCESS in 1h 21m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ca9a970f82f84d6eabf39d2c70544072 : SUCCESS in 2h 05m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d8137a27bfc14613be57221ba0461c3f : SUCCESS in 1h 30m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bec65554830c4c7fada1a4db6d2899c9 : SUCCESS in 1h 07m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cab4595d4f444521977d0ecba8ecb338 : TIMED_OUT in 3h 03m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3165c3adb6a14c53a65e0a4d6efec028 : SUCCESS in 2h 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ea96614e4d4d3baf6e81d9e671c650 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59e41ed905594e758c879261a43e2f8d : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"90a7f3d06e9d879b1ca2cf08f980779666e55798","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-21 09:20:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"140fb13a6bacfd4b029cd182f54dad096dd40345","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 12:24:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ce947f93bae8447db4cae1090a91ed94 : SUCCESS in 5m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/1346c02a917d44f79fb2007a2ddc5923 : SUCCESS in 1h 17m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/af569eace08445dea94849d8c2e18cc3 : SUCCESS in 1h 54m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a6fff843d893468193d7798ccf2c0758 : SUCCESS in 1h 28m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2bfa9504363a464b8e8aa34317fc456d : SUCCESS in 1h 16m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/40e10afe43e04dcea1b50dcaeb3f4784 : TIMED_OUT in 3h 03m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9980924521944bffa229f7b86752b3f8 : FAILURE in 1h 08m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adb61dbe58554b3dab4b073b1f3989af : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc8ae53847724003a7b4066433aa4579 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"378dded86785544145a51b20f7401a304511bbd0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-21 20:17:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9242907e39dee7ac9ef8616cd19b8e4c4a809da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-21 23:13:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6c564248c7084f27b56d89550421125f : SUCCESS in 4m 18s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/2c09addbdd6e41818b223e00bf261ce5 : SUCCESS in 1h 17m 31s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3bdd09d072494a42a0a1c52b05f84c69 : SUCCESS in 1h 40m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a52fa13c04704629837856a2b7bb6779 : SUCCESS in 1h 28m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7058884bac054af6b967e7cb7f1b8ae4 : SUCCESS in 1h 11m 23s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/79c363d2f142499d9faded354c8498d2 : SUCCESS in 2h 54m 50s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fef27ae6840848a2a40d45b88c2f3d2b : SUCCESS in 2h 01m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3260342c764b4077a7021f744131f7a9 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7826c3f012c489bb3f777a07609cc88 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"fce19f12f72430f0361d48f49277ef92a8c68afd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-22 07:56:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1a4054a2d8ec7d8ccff22958d67350415dbd564e","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-11-22 13:22:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d654281d9cb629718f640729b89d355b41509da6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 13:22:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0957ec4bc259d2be133f1d2746cadac4a4f8136f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 16:17:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"acd90611b74c03e10c57e9da0ab8313dfb10bd61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 16:17:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b1be831a8f604b1391a403cc7ec37938 : SUCCESS in 8m 29s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f9d180d759a543988308789f29070e73 : SUCCESS in 1h 46m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b155b694d8fe46bd9938849ef7f46f31 : SUCCESS in 1h 37m 53s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/efd64cf85b2b4777bf1e4ebaf57d8711 : SUCCESS in 1h 34m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43d26ab60963435abec90ea91c38017f : SUCCESS in 1h 29m 47s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f450016aa88b4034a074a049bd8884fb : SUCCESS in 2h 51m 44s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3dbb4de136be4d7e87fec89fc1603482 : SUCCESS in 2h 24m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d05b43ed48348af82484ae56afd9c67 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/109b77b279944a87bb5e65d2dad63454 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"af803e92d14009fb5f7e9ec50198e6cfb921ad91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 16:18:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e24a82c22e4185aff49a8bb74651e6 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e975950dcac4483b7b1d5634cb71847 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18e1b6c6513630f5543893ad5c07c7a06aecba67","revisions":{"18e1b6c6513630f5543893ad5c07c7a06aecba67":{"kind":"REWORK","_number":1,"created":"2019-11-07 03:13:41.000000000","uploader":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"ref":"refs/changes/81/693281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/81/693281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/81/693281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/81/693281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/81/693281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/81/693281/1"}}},"commit":{"parents":[{"commit":"122fa765b3aa67c520babfbc5f42f689d7762a39","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/122fa765b3aa67c520babfbc5f42f689d7762a39"}]}],"author":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-11-07 03:13:09.000000000","tz":480},"committer":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-11-07 03:13:38.000000000","tz":480},"subject":"Fix typos in yml files","message":"Fix typos in yml files\n\nChange-Id: I9e0c48a6d6a6e6b128add994e4df075b334b8984\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/18e1b6c6513630f5543893ad5c07c7a06aecba67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/18e1b6c6513630f5543893ad5c07c7a06aecba67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
