)]}'
{"id":"openstack%2Fopenstack-ansible-os_ironic~697709","triplet_id":"openstack%2Fopenstack-ansible-os_ironic~stable%2Ftrain~I2ee09d9cc20f9b8e4430c55129cd8bac9435299d","project":"openstack/openstack-ansible-os_ironic","branch":"stable/train","topic":"osa-train","hashtags":[],"change_id":"I2ee09d9cc20f9b8e4430c55129cd8bac9435299d","subject":"Adds configuration for the Ironic Inspector","status":"MERGED","created":"2019-12-06 19:24:27.000000000","updated":"2019-12-07 08:47:59.000000000","submitted":"2019-12-07 08:45:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697709-1575708359497-82587bb4","meta_rev_id":"fad5529a34c6217cbd3ffbec24462706420931ec","_number":697709,"virtual_id_number":697709,"owner":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-06 22:33:20.000000000","_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-07 08:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":2,"date":"2019-12-07 02:24:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-12-07 06:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":1,"date":"2019-12-07 06:48:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-07 02:24:44.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-12-07 06:48:49.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-12-07 08:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43622518db1c17081df2adfd7d1163c28a490c9a","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-12-06 19:24:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1a27f45778d92a03a0a2e83a764fcc4014f15270","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-12-06 19:25:02.000000000","message":"Topic changed from osa-train-stable/train to osa-train","accounts_in_message":[],"_revision_number":1},{"id":"e458ba71e64b32692da6de8f240dc41dce6e8584","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-12-06 21:51:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a1a1a6635f96cdf6b9c02ad5f01591cc30ec57c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 22:32:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b9fad90a8a0448348f9b9a88a4156451 : SUCCESS in 6m 19s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5e77454794b249c1a51aea8fe6e4cd58 : FAILURE in 1h 21m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/95deb155001e448d977b4c4264748c45 : SUCCESS in 1h 54m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e240ead8aec84ae3af14da531011defc : SUCCESS in 1h 31m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8684efcaf44140018fd2e2fd12d3650d : FAILURE in 1h 16m 08s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/56d9235476d442c38a2f9e9e779b5867 : SUCCESS in 3h 02m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b3022d68293e40b8ad0b1afde6cdf54e : SUCCESS in 2h 21m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea63474b1635444aaffc0b6a079171f4 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5f59ea17ff74fd89c9e7c26cb5599bc : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d5318b3c2daf1e5dc2c9dd078cc352497c2893e3","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-12-06 22:33:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0760867acee4a2cf1449ec08221082ac138fbe9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 01:41:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/baa9024eccb64ddca77db2723f50d5df : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b25f19ea158345b1bb4331be35a3a768 : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e060293260fa4196ac0084256ec13c96 : SUCCESS in 2h 03m 47s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e4ad8dff0e8a455b83a9b7b1fc630a24 : SUCCESS in 1h 20m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b390f60aec744359a15f359e88d66481 : SUCCESS in 1h 18m 55s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/85c1c222167a464ea1ee9d0abde45982 : SUCCESS in 3h 02m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a6882edc824414c877e69390ef7bee0 : SUCCESS in 1h 58m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dc1e785e91c458890067e4ab6aac821 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e89f50b5f8a446e860021fc51c3ac45 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":1},{"id":"f4cd4bd6dd26de38c3b225d05d1c715073a36f35","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-12-07 02:24:44.000000000","message":"Patch Set 1: Code-Review+2\n\ngood stuff","accounts_in_message":[],"_revision_number":1},{"id":"c747660e5af8f9546a24df879dc3e4b08bf73745","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-07 06:48:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4f0743e1ef05e34343498120f59e7a9799527c79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 06:49:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d2e70c75a46fd4de4bc115732c6324e458a0e52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 08:45:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1a246e034cc501bb9437ab7c4b062353e5d03a9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 08:45:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/92d00ff7513448329a701186a23d6b63 : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/44d5dedb2c804638b646a9fd9f05cfc8 : SUCCESS in 1h 17m 54s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/974e94c3dd9148cebd890d515c7162c6 : SUCCESS in 1h 50m 50s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b0eea29ab8a040ab8a346d3cdfa539e9 : SUCCESS in 1h 33m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1616c9c54c07470eb45425509b985742 : SUCCESS in 1h 27m 43s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a4e61c2363141a89664f795beae0a81 : SUCCESS in 1h 53m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d12fdaf259384ad2a80d2722bf7474a9 : SUCCESS in 17m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/193b79eb268c4aaa8010934eda73ca5e : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"fad5529a34c6217cbd3ffbec24462706420931ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-07 08:47:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7adc8c9f06a5488da116c35f025c064d : SUCCESS in 1m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1fbc21c03e048c39b747b4f6b36d587 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"209e2465af340ecfdcc163d2b619e7d78ee3a0c8","revisions":{"209e2465af340ecfdcc163d2b619e7d78ee3a0c8":{"kind":"REWORK","_number":1,"created":"2019-12-06 19:24:27.000000000","uploader":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"ref":"refs/changes/09/697709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/09/697709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/09/697709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/09/697709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/09/697709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/09/697709/1"}}},"commit":{"parents":[{"commit":"ba59391eca583d5608df19c5aa1ed604fbdc6d7d","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/ba59391eca583d5608df19c5aa1ed604fbdc6d7d"}]}],"author":{"name":"Cam Loader","email":"cameron@loaderusa.net","date":"2019-09-05 02:36:30.000000000","tz":-300},"committer":{"name":"Cam J. Loader (cjloader)","email":"cameron@loaderusa.net","date":"2019-12-06 19:24:27.000000000","tz":0},"subject":"Adds configuration for the Ironic Inspector","message":"Adds configuration for the Ironic Inspector\n\nThis commit enables and configures the Ironic Inspector. This feature\nallows for baremetal nodes to be introspected. This provides useful\ninformation about an Ironic host. Such information includes harware\nand mac addresses.\n\nDepends-On: https://review.opendev.org/680553\nChange-Id: I2ee09d9cc20f9b8e4430c55129cd8bac9435299d\n(cherry picked from commit 7da7127071a33d2a5dcb750592c1fb9854bdbebb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/209e2465af340ecfdcc163d2b619e7d78ee3a0c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/209e2465af340ecfdcc163d2b619e7d78ee3a0c8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
